)]}'
{"id":"openstack%2Fvenus~823118","triplet_id":"openstack%2Fvenus~master~Ie10077d2f2f1be0423b9eeb0d5da9d84aba50c4b","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2021-12-29 02:33:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie10077d2f2f1be0423b9eeb0d5da9d84aba50c4b","subject":"add designate log input config","status":"MERGED","created":"2021-12-29 01:59:40.000000000","updated":"2021-12-29 02:33:58.000000000","submitted":"2021-12-29 02:33:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823118","meta_rev_id":"db30ba17a9194cc9e076290c4365d194348108ba","_number":823118,"virtual_id_number":823118,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-29 02:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2021-12-29 02:16:53.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2021-12-29 02:16:53.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-29 02:13:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-29 02:16:53.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"1c02cca0f4c1fb6e624250c7526e3a748d25718a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2021-12-29 01:59:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e4c4b1cafc373b3dd7598eea4a89540b6b7946e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 02:13:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c0215cb06147ab8326f463cd88a2b4 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60704a513ff1474682a3077fe839d6df : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c41999c8164e415fb4094a11f569fa91 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954fe96af0704f2eb63fcd1d36c17081 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f2854d1567ca0d3a92a5776f1d853bed571f3e21","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2021-12-29 02:16:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e969fb53ee01d4ea747ed151f3ffa7ee4b5735b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 02:21:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4c9ecfd8800f073615cf48049181732779dc47f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 02:33:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc1791265ee407d9162e959bc71153e : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5616a825e184e2bb65157e35b350c9b : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e488fd123cb840a58aec373fabf5948f : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b4e2f8357a47d0a05711609b310ff0 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"db30ba17a9194cc9e076290c4365d194348108ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 02:33:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1379ec8e220314d725eca623b40476b93c08b17c","revisions":{"1379ec8e220314d725eca623b40476b93c08b17c":{"kind":"REWORK","_number":1,"created":"2021-12-29 01:59:40.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/18/823118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/18/823118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/18/823118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/18/823118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/18/823118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/18/823118/1"}}},"commit":{"parents":[{"commit":"0f19dfe297dd493b40454f92f8b688f6c11af297","subject":"add octavia log input config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/0f19dfe297dd493b40454f92f8b688f6c11af297"}]}],"author":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2021-12-29 01:59:27.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2021-12-29 01:59:27.000000000","tz":480},"subject":"add designate log input config","message":"add designate log input config\n\nChange-Id: Ie10077d2f2f1be0423b9eeb0d5da9d84aba50c4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1379ec8e220314d725eca623b40476b93c08b17c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1379ec8e220314d725eca623b40476b93c08b17c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
