)]}'
{"id":"openstack%2Fvenus~823154","triplet_id":"openstack%2Fvenus~master~I33c6c546bebc88eeafea47c99b0d2d6c8dae6deb","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-01-05 00:54:48.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-01-03 06:18:33.000000000","reason":"Brin Zhang replied on the change"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-01-05 00:54:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I33c6c546bebc88eeafea47c99b0d2d6c8dae6deb","subject":"Dict get properties default is None","status":"MERGED","created":"2021-12-29 08:54:02.000000000","updated":"2022-01-05 00:57:20.000000000","submitted":"2022-01-05 00:54:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823154","meta_rev_id":"326155a9d42b6927c28b459427f6f4160b5c8146","_number":823154,"virtual_id_number":823154,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-05 00:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"all":[{"value":2,"date":"2022-01-05 00:43:59.000000000","_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},{"value":2,"date":"2022-01-03 06:18:33.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"all":[{"value":1,"date":"2022-01-05 00:43:59.000000000","_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-29 09:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-02 03:23:59.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-01-02 03:23:59.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-01-05 00:43:59.000000000","updated_by":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"reviewer":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"state":"REVIEWER"}],"messages":[{"id":"96fed8652dde79539cfd008a118dc1d4ffd7fcf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-12-29 08:54:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82cbbea087e55308d12b667e09896ad1cf10ffc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-29 09:02:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12530d9e3a1d4d1b8d4a34b60a07e271 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1684fbd114d74e25a369a0b1c4800e8d : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc7fb4dc68494082be152cf1ebc34256 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c54469e713f4f56a05f9d886d7203f4 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e70576cce056d5302f347ec3205d84e0f00e5abe","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-01-03 06:18:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85767dd52d89aa030da83ce08f598ca27244ee79","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2022-01-05 00:43:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"904361f2e43c736ac31f9dd79b672ae01c48d9ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 00:44:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"988edb79eaf0a61246e7ac3102ed5937ce9436c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 00:54:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79066b0b8e7c4f68b9e338b66df0023e : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ede599d06134cc0b6dec1484a60294d : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e06c8fa5978a487cad9c131a867b7472 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd176a7b73c4d8ca4550a33eeec9739 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"19afd52695a5dd6f76faddccd57dc8dee4b22d0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 00:54:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"326155a9d42b6927c28b459427f6f4160b5c8146","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 00:57:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed519a38a4c7494494d6196d35aee995 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28deed28387d13bc0448d2a1c212484d8fc14ff9","revisions":{"28deed28387d13bc0448d2a1c212484d8fc14ff9":{"kind":"REWORK","_number":1,"created":"2021-12-29 08:54:02.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/54/823154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/54/823154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/54/823154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/54/823154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/54/823154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/54/823154/1"}}},"commit":{"parents":[{"commit":"77e55cda7e04eec93c33bad473f7c500b06df292","subject":"Add unit test for params","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/77e55cda7e04eec93c33bad473f7c500b06df292"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-12-29 08:53:51.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-12-29 08:53:51.000000000","tz":0},"subject":"Dict get properties default is None","message":"Dict get properties default is None\n\nChange-Id: I33c6c546bebc88eeafea47c99b0d2d6c8dae6deb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/28deed28387d13bc0448d2a1c212484d8fc14ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/28deed28387d13bc0448d2a1c212484d8fc14ff9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
