)]}'
{"id":"openstack%2Fvenus~824515","triplet_id":"openstack%2Fvenus~master~I65e8f8de096a73a10053246d017d24806db8f160","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34017":{"account":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"last_update":"2022-01-13 09:22:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I65e8f8de096a73a10053246d017d24806db8f160","subject":"Fix manage to manager","status":"MERGED","created":"2022-01-13 09:01:38.000000000","updated":"2022-01-13 09:23:53.000000000","submitted":"2022-01-13 09:22:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824515","meta_rev_id":"6981d3037d7bb209ae4fd15c1bc89d0380917f38","_number":824515,"virtual_id_number":824515,"owner":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-13 09:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2022-01-13 09:05:32.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2022-01-13 09:05:32.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-13 09:05:32.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2022-01-13 09:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8efb5f91c1eb0e9e1e5b75e1ea8d97a6e080d51d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-13 09:01:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67ca01662bf0bc04b1cf18b94ef35e49698254e7","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-01-13 09:05:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"127b164df93560cb4908c73b1184057e7c1f0166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:14:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b6d3f992cfc241349fee8264a38f42c3 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/304ac2f513794b66931f79c3f9322214 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5cd69cfe6224cccb757541e233ff50b : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c0289d7cca44207829a9533a98e7589 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5042f73099d40beb9b2de2d96ac0ff7 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ba3fabaeb10c3a3fb025fb690e2e500fb9438fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:14:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a09296b8fc72c8abf50c2c9f753ac2a54d1054de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:22:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/80e71e8097af48e48e2e718340883cae : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73abaf4104b46839519ffe2523fb85d : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca79fb802a454a6a93a325c7b7e2a87f : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adb720202ffd41d280486867361e6efa : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4d8faded34e40d3b4a687589e9554e3 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"73d149738fdbe1a7c18b49c8ffa0ad10aa9b0e63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:22:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6981d3037d7bb209ae4fd15c1bc89d0380917f38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:23:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9870a42cb57341978ff2cf52c1067a23 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9fad2b2f39fca327707dacde66e21c054cc45ba","revisions":{"b9fad2b2f39fca327707dacde66e21c054cc45ba":{"kind":"REWORK","_number":1,"created":"2022-01-13 09:01:38.000000000","uploader":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"ref":"refs/changes/15/824515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/15/824515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/824515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/824515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/824515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/15/824515/1"}}},"commit":{"parents":[{"commit":"96903af56deba3721fe4935cd52f3ec804b6acf2","subject":"Merge \"Dict get properties default is None\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/96903af56deba3721fe4935cd52f3ec804b6acf2"}]}],"author":{"name":"jinyuanliu","email":"jayonlau@gmail.com","date":"2022-01-13 08:59:43.000000000","tz":-300},"committer":{"name":"jinyuanliu","email":"jayonlau@gmail.com","date":"2022-01-13 09:01:34.000000000","tz":-300},"subject":"Fix manage to manager","message":"Fix manage to manager\n\nOtherwise, relevant commands are missing in the container\n\nChange-Id: I65e8f8de096a73a10053246d017d24806db8f160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b9fad2b2f39fca327707dacde66e21c054cc45ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b9fad2b2f39fca327707dacde66e21c054cc45ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
