)]}'
{"id":"openstack%2Fvenus~824663","triplet_id":"openstack%2Fvenus~master~Ia4d9952e5687fed63ced98ca2803358180aea139","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34017":{"account":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"last_update":"2022-01-14 07:51:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4d9952e5687fed63ced98ca2803358180aea139","subject":"add apscheduler","status":"MERGED","created":"2022-01-14 01:42:54.000000000","updated":"2022-01-14 07:53:14.000000000","submitted":"2022-01-14 07:51:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824663","meta_rev_id":"8f580f589edfc60ba93f9d95519f9fb54e346c60","_number":824663,"virtual_id_number":824663,"owner":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-14 07:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2022-01-14 07:42:20.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2022-01-14 07:42:20.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 01:49:53.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2022-01-14 01:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79729e94bcf6bd63f8419acdd05537bc6baf0011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-14 01:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19d4d3e2dc566d30ed953a61a2e495a31d2f0271","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-01-14 01:49:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e830adbca1c981122e0bb45474ab543f16c9cf9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 01:50:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aef2f79afead46de918980f1986a6034 : FAILURE in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/947a384c496c40609c12fc7c1f5ad271 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4158fe987ca4d99bfadfcec3fad8269 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcfd79fd8e6843f594231cb71d326974 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c6592a42714c289d7e66d00740cd7f : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"519ff64c3f15ba72272d784d5d626ee063487947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-14 07:16:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4ec4649082598b7ef6d1dfe13c0dbde3c629f300","author":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"date":"2022-01-14 07:17:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e412ad3e9d2fabc24df99b9c4cfb88be73dcb8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 07:27:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d60c3c83fcee4a28b68935daeb0c439c : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea047c0f627744d0adeba9981a24dee2 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98cc5e15561b45aabafe71a40b659463 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/375d110912234b40adf630d8ab40a38f : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f88628e6738b4301b225f1e729f1ecb0 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"75fb3d074f99320232899db30dcda60189329388","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-01-14 07:42:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d79d3e6c245f026d3640160b5ab9ea06f30c93f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 07:42:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d2504d2e79e1f2eb2e6785d33d5301032b84817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 07:51:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2da77f82f0374c4f838d1546402ea1d5 : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be62b7e07f7743d1b5f42bf252db66dc : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/920b7f6b01b54bf2899eee113c77c1a5 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d7f0687f80b43e4be752d53e8c85e38 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409ae9d031074934b74bb6a7c29c992c : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"733ffa982026113b860664fda2ec81adc9f3fa76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 07:51:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8f580f589edfc60ba93f9d95519f9fb54e346c60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 07:53:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c8cdf5032d407d975ee72248a47d7c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a351c3ac51d1fc225338b52571aa8302bddc7dca","revisions":{"2faad053ba5353524c9c1d907f19047f9a1ad1d6":{"kind":"REWORK","_number":1,"created":"2022-01-14 01:42:54.000000000","uploader":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"ref":"refs/changes/63/824663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/63/824663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/824663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/824663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/824663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/63/824663/1"}}},"commit":{"parents":[{"commit":"b9fad2b2f39fca327707dacde66e21c054cc45ba","subject":"Fix manage to manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b9fad2b2f39fca327707dacde66e21c054cc45ba"}]}],"author":{"name":"jinyuanliu","email":"jayonlau@gmail.com","date":"2022-01-14 01:41:27.000000000","tz":-300},"committer":{"name":"jinyuanliu","email":"jayonlau@gmail.com","date":"2022-01-14 01:42:51.000000000","tz":-300},"subject":"add apscheduler","message":"add apscheduler\n\nadd spscheduler for MIT License\n\nChange-Id: Ia4d9952e5687fed63ced98ca2803358180aea139\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2faad053ba5353524c9c1d907f19047f9a1ad1d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2faad053ba5353524c9c1d907f19047f9a1ad1d6"}]},"branch":"refs/heads/master"},"a351c3ac51d1fc225338b52571aa8302bddc7dca":{"kind":"REWORK","_number":2,"created":"2022-01-14 07:16:56.000000000","uploader":{"_account_id":34017,"name":"jayonlau","display_name":"jinyuanliu","email":"jayonlau@gmail.com","username":"lau"},"ref":"refs/changes/63/824663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/63/824663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/824663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/824663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/824663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/63/824663/2"}}},"commit":{"parents":[{"commit":"b9fad2b2f39fca327707dacde66e21c054cc45ba","subject":"Fix manage to manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b9fad2b2f39fca327707dacde66e21c054cc45ba"}]}],"author":{"name":"jinyuanliu","email":"jayonlau@gmail.com","date":"2022-01-14 01:41:27.000000000","tz":-300},"committer":{"name":"jayonlau","email":"jayonlau@gmail.com","date":"2022-01-14 07:16:55.000000000","tz":0},"subject":"add apscheduler","message":"add apscheduler\n\nadd spscheduler for MIT License\n\nChange-Id: Ia4d9952e5687fed63ced98ca2803358180aea139\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a351c3ac51d1fc225338b52571aa8302bddc7dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a351c3ac51d1fc225338b52571aa8302bddc7dca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
