)]}'
{"id":"openstack%2Fvenus~844428","triplet_id":"openstack%2Fvenus~master~Ie16aa50341f3ffffa2fd45d51af58d4f7a8b9813","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2022-06-02 12:19:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie16aa50341f3ffffa2fd45d51af58d4f7a8b9813","subject":"collect heat logs from defalut location of /var/log","status":"MERGED","created":"2022-06-02 12:04:39.000000000","updated":"2022-06-02 12:21:00.000000000","submitted":"2022-06-02 12:19:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844428","meta_rev_id":"f79338125345e8315e1f4ffba27a3b4d16ae5156","_number":844428,"virtual_id_number":844428,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 12:19:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2022-06-02 12:12:29.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2022-06-02 12:12:29.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 12:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 12:12:29.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"7e24a7b3b5e046b5ae8e446fefa30a85dcb0d0a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-06-02 12:04:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0282bc0fcac0670f30527357eddb5a664458329b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 12:12:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fa4027f7484f15b8aa0eb426461411 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9abb3835cc34855a47148d60a997424 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6a5ecd9af28355df77f7911c0295450557473185","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2022-06-02 12:12:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"147d9fa2027a15062107b71ea351bd9e23cadf52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 12:12:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73d6f124d4d1e88d52c11c842597236486f16bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 12:19:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef5bcc17ae44e0b9fdfc3832b6beb9b : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77556c50dd664742b9be98c7cda9d71f : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f257f4ca741f738308c7f984519b2119e69a67b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 12:19:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f79338125345e8315e1f4ffba27a3b4d16ae5156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 12:21:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ba3e6cdb304126858c03fe9a778819 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2714a46452a7f65fec9e4990c31c5038346ff06f","revisions":{"2714a46452a7f65fec9e4990c31c5038346ff06f":{"kind":"REWORK","_number":1,"created":"2022-06-02 12:04:39.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/28/844428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/28/844428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/28/844428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/28/844428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/28/844428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/28/844428/1"}}},"commit":{"parents":[{"commit":"fcb11d9aac8add9bb6434eb5290154a728cee846","subject":"Merge \"use system logdir /var/log/ for collect glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/fcb11d9aac8add9bb6434eb5290154a728cee846"}]}],"author":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2022-06-02 12:04:23.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2022-06-02 12:04:23.000000000","tz":480},"subject":"collect heat logs from defalut location of /var/log","message":"collect heat logs from defalut location of /var/log\n\nChange-Id: Ie16aa50341f3ffffa2fd45d51af58d4f7a8b9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2714a46452a7f65fec9e4990c31c5038346ff06f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2714a46452a7f65fec9e4990c31c5038346ff06f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
