)]}'
{"id":"openstack%2Fvenus~849014","triplet_id":"openstack%2Fvenus~master~I50443af0aa040cee228b539dbea2ba4b990e7d73","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-07-11 02:35:31.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-07-14 00:42:50.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-07-14 00:34:31.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-07-14 00:42:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50443af0aa040cee228b539dbea2ba4b990e7d73","subject":"Add a test cases for unit tests","status":"MERGED","created":"2022-07-08 11:25:24.000000000","updated":"2022-07-14 00:43:45.000000000","submitted":"2022-07-14 00:42:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849014","meta_rev_id":"7adde31cdb1d6e16807f9de9b2a5d8ded82236dd","_number":849014,"virtual_id_number":849014,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 00:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-07-11 06:07:19.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-07-14 00:34:31.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-07-14 00:34:31.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 11:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 01:58:00.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2022-07-11 02:35:46.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-07-11 02:35:46.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"39c013e0d29756763ee874818e3beb1cc9dbeb56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-07-08 11:25:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a718f87a303408f31ab27983e4aab7cbca1192b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 11:31:51.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1ed84129ac740cd83731f304c30d26b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd182b57505c4cbd91d7d04036816b9f : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42e7a3da7cd24225b4d5a028d42e336e : FAILURE in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f8a60014e8e4f2d954f557fb7c2bb79 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9cc0e8fe760f3671d24129135d69aaadf75f4bda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-07-11 01:56:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"091651c3cc11f14ed9250a7080de0eeda2428b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 02:04:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14aac7bbc3c545128a75b8a7c6174138\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5b4a3ffdf64737add2b9bfc1a33835 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa320207e5b44a9aa5c72edb6468bb8 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f38c8e94516b49f99ed1ddda32ede28f : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"521145c3b333f87d258f6a82ea605f4d6fa0c75b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-07-11 02:35:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"06bb70a0c7b9d93bffb205fae3962e049f7de38a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-07-11 02:51:06.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3a81a8af28f2dc3a96e5d5fa8b7d5d65eae8ffaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 02:58:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa24d1d9b55a4d2686fc84dfdef87370\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b60ef0c644e44359cad020344ef596e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dfae8231ffe4d3cb5ea1ce8ae48a485 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76bd613614a247f082007a48346586ae : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"408b42a34e1f7432537494b057e02a886ea14a6e","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-07-11 06:07:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e02939c8d449a553b658af5f52a55755d86dbca1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-07-14 00:34:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"93aab2a0f42e70e451b176e531dfe46c2fece610","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 00:34:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f81fba85eab5262dde8f9b797b36c4680a95a247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 00:42:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b414cd59de4cc9b57e584aea76268f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364053864eca47c388f5056778a01ade : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/facef99162b7428eb7e5cdb8f20e7e96 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8686c4c71ad421d9fbb5b99894addc8 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"0cbf67d047c53a410ecdcec186515f3beace500a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 00:42:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7adde31cdb1d6e16807f9de9b2a5d8ded82236dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 00:43:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b84562f5b4694abbbc14eedf1ef060c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de72a3b1611349d3bff0e556f2f9d7ab : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e16dcbd9899fc4ff0178dc33b6138a126f8c49f2","revisions":{"43e67605938003303eab6131be52bb5a77e3b669":{"kind":"REWORK","_number":1,"created":"2022-07-08 11:25:24.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/14/849014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/14/849014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/14/849014/1"}}},"commit":{"parents":[{"commit":"7c3ec56142620cb2b3ff1562947c65a266401d57","subject":"Merge \"Add test cases to unit tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7c3ec56142620cb2b3ff1562947c65a266401d57"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-08 11:20:56.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-08 11:25:18.000000000","tz":480},"subject":"Add a test cases for unit tests","message":"Add a test cases for unit tests\n\nAdd A test case to A so that if there is A lowercase character\nin the input, it will be uppercase in the output.\n\nChange-Id: I50443af0aa040cee228b539dbea2ba4b990e7d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/43e67605938003303eab6131be52bb5a77e3b669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/43e67605938003303eab6131be52bb5a77e3b669"}]},"branch":"refs/heads/master"},"e563cdc584e12715cdc3792f28dd40900e8029bf":{"kind":"REWORK","_number":2,"created":"2022-07-11 01:56:10.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/14/849014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/14/849014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/14/849014/2"}}},"commit":{"parents":[{"commit":"7c3ec56142620cb2b3ff1562947c65a266401d57","subject":"Merge \"Add test cases to unit tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7c3ec56142620cb2b3ff1562947c65a266401d57"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-08 11:20:56.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-11 01:55:49.000000000","tz":480},"subject":"Add a test cases for unit tests","message":"Add a test cases for unit tests\n\nAdd A test case to A so that if there is A lowercase character\nin the input, it will be uppercase in the output.\n\nChange-Id: I50443af0aa040cee228b539dbea2ba4b990e7d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e563cdc584e12715cdc3792f28dd40900e8029bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e563cdc584e12715cdc3792f28dd40900e8029bf"}]},"branch":"refs/heads/master"},"e16dcbd9899fc4ff0178dc33b6138a126f8c49f2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-11 02:51:06.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/14/849014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/14/849014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/849014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/14/849014/3"}}},"commit":{"parents":[{"commit":"7c3ec56142620cb2b3ff1562947c65a266401d57","subject":"Merge \"Add test cases to unit tests.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7c3ec56142620cb2b3ff1562947c65a266401d57"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-08 11:20:56.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-11 02:51:06.000000000","tz":0},"subject":"Add a test cases for unit tests","message":"Add a test cases for unit tests\n\nAdd a test case to \u0027generate_must\u0027 so that if there are lowercase characters\nin the input, it will be uppercase in the output.\n\nChange-Id: I50443af0aa040cee228b539dbea2ba4b990e7d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e16dcbd9899fc4ff0178dc33b6138a126f8c49f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e16dcbd9899fc4ff0178dc33b6138a126f8c49f2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[]}
