)]}'
{"id":"openstack%2Fvenus~849693","triplet_id":"openstack%2Fvenus~master~I51212cb48ef7f94c09445469aa4025430e9375a3","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-07-14 05:37:51.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-07-14 06:33:35.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-07-14 06:26:37.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I51212cb48ef7f94c09445469aa4025430e9375a3","subject":"Supplement test cases for test_generate_must.","status":"MERGED","created":"2022-07-14 02:43:32.000000000","updated":"2022-07-14 06:34:30.000000000","submitted":"2022-07-14 06:33:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849693","meta_rev_id":"58377e013a2187a95fd2d23e9ff1757b61372184","_number":849693,"virtual_id_number":849693,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 06:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-07-14 05:37:51.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-07-14 06:26:37.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-07-14 06:26:37.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-14 02:45:32.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-07-14 02:45:32.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2022-07-14 02:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1134e895004df1da0c32d7fcc760a9997d6eec44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-07-14 02:43:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"325e3638806f15b3447de1cc1faeb45a072a9464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 02:58:36.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f38c1e00b76486eae0608c34e190223\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed6e89657ae4e73a10966cfe7d48b9d : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17de79ad68e242c492fcb47bca38a0d8 : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5dcd6171092469db0900a8410f0d359 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8da7334814dc270ee86beb5d2a5f3efd5bdae341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-07-14 03:32:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9cdf4ea3a501241dcd98bae4c421dec4ed0fb4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 03:38:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2675693bb50d4008ade8f2a0b964a36a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e063d51e75984845af6dd227b82a588f : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345b06c0964c48478757cf9899a626f7 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/721f37a3b035451098cd59eeae59136a : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2f57a1a2417b892e6b82ffe60d4de9dc84aed7af","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-07-14 05:37:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"645ac12ae30af27a48f37ad6812ad0832b4e0e2b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-07-14 06:26:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22ce3f11d234599bd3df05683f2c237edd456836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 06:26:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9664a3d87db720c86e49a3c41df49ee8b00cf7a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 06:33:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e19ac0e4d9a2483bb66abef4aca00466\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a414d5fd7914f4b95727d058f0b4ac0 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03917027d13f47069c2d63f5f13338eb : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25f1e3a595794fffac05b0fd21917e18 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0d861e0b6d3518565eb349aa4fbe41ab5adaee84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 06:33:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"58377e013a2187a95fd2d23e9ff1757b61372184","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 06:34:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ec776f7d4c434bb1eaa3bc91948205\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/890335a78f3e405694c5fc0cbcb09735 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28635ed6f114cc5e909abd0180a62e8c24493066","revisions":{"5016a2efed4ed5e933e55dbe1e1a606c60cd64c1":{"kind":"REWORK","_number":1,"created":"2022-07-14 02:43:32.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/93/849693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/93/849693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/93/849693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/93/849693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/93/849693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/93/849693/1"}}},"commit":{"parents":[{"commit":"e16dcbd9899fc4ff0178dc33b6138a126f8c49f2","subject":"Add a test cases for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e16dcbd9899fc4ff0178dc33b6138a126f8c49f2"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-14 02:40:17.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-14 02:43:27.000000000","tz":480},"subject":"Supplement test cases for test_generate_must.","message":"Supplement test cases for test_generate_must.\n\nIn test_generate_must_lower, if the input is uppercase, the output\nwill change them to lowercase. If k is not \u0027keyword\u0027,\ntest_generate_must_other is entered.\n\nChange-Id: I51212cb48ef7f94c09445469aa4025430e9375a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5016a2efed4ed5e933e55dbe1e1a606c60cd64c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5016a2efed4ed5e933e55dbe1e1a606c60cd64c1"}]},"branch":"refs/heads/master"},"28635ed6f114cc5e909abd0180a62e8c24493066":{"kind":"REWORK","_number":2,"created":"2022-07-14 03:32:12.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/93/849693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/93/849693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/93/849693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/93/849693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/93/849693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/93/849693/2"}}},"commit":{"parents":[{"commit":"e16dcbd9899fc4ff0178dc33b6138a126f8c49f2","subject":"Add a test cases for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e16dcbd9899fc4ff0178dc33b6138a126f8c49f2"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-14 02:40:17.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-07-14 03:31:53.000000000","tz":480},"subject":"Supplement test cases for test_generate_must.","message":"Supplement test cases for test_generate_must.\n\nIn test_generate_must_lower, if the input is uppercase, the output\nwill change them to lowercase. If k is not \u0027keyword\u0027,\ntest_generate_must_other is entered.\n\nChange-Id: I51212cb48ef7f94c09445469aa4025430e9375a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/28635ed6f114cc5e909abd0180a62e8c24493066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/28635ed6f114cc5e909abd0180a62e8c24493066"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[]}
