)]}'
{"id":"openstack%2Fvenus~851635","triplet_id":"openstack%2Fvenus~master~I483247f6b092f0dea5d58dfaee12e7fd9cc03651","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-08-02 01:13:32.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-08-10 09:25:50.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-08-08 02:37:15.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I483247f6b092f0dea5d58dfaee12e7fd9cc03651","subject":"Add a test cases for \u0027logs\u0027","status":"MERGED","created":"2022-08-01 11:43:53.000000000","updated":"2022-08-10 09:26:51.000000000","submitted":"2022-08-10 09:25:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851635","meta_rev_id":"d1762531825e86a152dcb2adb9cd6d356270c219","_number":851635,"virtual_id_number":851635,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 09:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-08-10 09:13:55.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-08-08 02:37:15.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2022-08-10 09:13:55.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-08-09 01:11:35.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-01 11:49:22.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-08-01 11:49:22.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2022-08-01 11:49:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"752bf14b1420ce238b9d39cc690f900120149213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-01 11:43:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9715eb9250e727dabb3e147b1b5f410041d9828e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 11:49:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ba29eba3794af1b2a5121627ddc62a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43bd1846ee6d411fa8271a4b1871bd27 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c32b7dce7804b4ca5bd5e2cbaf55017 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25df558206a84d9d9fc593c54348fbd1 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"116d52861de219d8085b1ed9c1aa832b471a5fcb","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-02 01:13:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d643441c865db6b865ffee248c486daa24a761e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-08 01:00:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7edad0b630460a1ed43311cccd384e6a94079634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 01:09:14.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95e0a12cf17743218606abbb34e56171\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873c08aec3754e11ade079eb29721071 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26689eff0e454130921dad0dc59f4783 : FAILURE in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce609806200045658610c766b826ccaf : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"8ab772b2c2804c1d29edff54ee53e9c7c6f8e558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-08 01:58:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"061a541d3444d582cc8b8826a2172f25e37d744b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 02:09:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74dfc7c806240bf8bcf2b3ba1f15e09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f57fa53969cb45c287e1f8f7bc20f8f2 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c7408057f54b28a6848e5c73872c6b : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91ec147e6cc64b8cac81a57860b666b6 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e7660804e80a2bbff6894508a9e7e034bf50fcfe","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-08 02:37:15.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8cbd18f64ad88cf4a2d02408de98ecce701f5678","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-09 01:11:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"589ee62204c7f12b08abbf93c01a729968573504","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-10 09:13:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3387f88de4ed10ab5329223dc8896c8f1b78d6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:15:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"496b8a20cf3b067fc1c6601d0e0ceec95804db8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:25:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7a22ade5aca49e18b3cf5599b014baa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/968294f449524339a5fb8784cbc5e0b2 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e25cfcfe5d64ddc9f1deeec6196737b : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a08919dc69a04284a82450a12b9a8140 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":3},{"id":"e13339c0b9d92dca8ca1cf22f275316e97d872fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:25:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d1762531825e86a152dcb2adb9cd6d356270c219","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:26:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6fee18638284354b42fb06662860738\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1338d2194244dad91cd663a5c8aefb2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ca570a8793bd15f2b976323de8883a3d4e61c828","revisions":{"aeb89951f2b3f3880744e1df6d8d982e9ffe7dc8":{"kind":"REWORK","_number":1,"created":"2022-08-01 11:43:53.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/35/851635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/35/851635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/35/851635/1"}}},"commit":{"parents":[{"commit":"d098d4a0d70a2415d3164c07c44fc66c917e2610","subject":"Another test case to supplement logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d098d4a0d70a2415d3164c07c44fc66c917e2610"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 11:40:58.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 11:40:58.000000000","tz":480},"subject":"Add a test cases for \u0027logs\u0027","message":"Add a test cases for \u0027logs\u0027\n\nAdd \u0027no data,no aggregations\u0027 test case of logs to\ntest_search_action.py.\n\nChange-Id: I483247f6b092f0dea5d58dfaee12e7fd9cc03651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/aeb89951f2b3f3880744e1df6d8d982e9ffe7dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/aeb89951f2b3f3880744e1df6d8d982e9ffe7dc8"}]},"branch":"refs/heads/master"},"7c180bd7425803434566fa1571e4dd58063845d9":{"kind":"REWORK","_number":2,"created":"2022-08-08 01:00:04.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/35/851635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/35/851635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/35/851635/2"}}},"commit":{"parents":[{"commit":"5b9ab46be3a25c35f1ec035b61cde713149f86ff","subject":"Add a test case of logs to the test file.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5b9ab46be3a25c35f1ec035b61cde713149f86ff"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 11:40:58.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-08 00:56:24.000000000","tz":480},"subject":"Add a test cases for \u0027logs\u0027","message":"Add a test cases for \u0027logs\u0027\n\nAdd \u0027no data,no aggregations\u0027 test case of logs to\ntest_search_action.py.\n\nChange-Id: I483247f6b092f0dea5d58dfaee12e7fd9cc03651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7c180bd7425803434566fa1571e4dd58063845d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7c180bd7425803434566fa1571e4dd58063845d9"}]},"branch":"refs/heads/master"},"ca570a8793bd15f2b976323de8883a3d4e61c828":{"kind":"REWORK","_number":3,"created":"2022-08-08 01:58:00.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/35/851635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/35/851635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/851635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/35/851635/3"}}},"commit":{"parents":[{"commit":"5b9ab46be3a25c35f1ec035b61cde713149f86ff","subject":"Add a test case of logs to the test file.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5b9ab46be3a25c35f1ec035b61cde713149f86ff"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 11:40:58.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-08 01:57:11.000000000","tz":480},"subject":"Add a test cases for \u0027logs\u0027","message":"Add a test cases for \u0027logs\u0027\n\nAdd \u0027no data,no aggregations\u0027 test case of logs to\ntest_search_action.py.\n\nChange-Id: I483247f6b092f0dea5d58dfaee12e7fd9cc03651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ca570a8793bd15f2b976323de8883a3d4e61c828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ca570a8793bd15f2b976323de8883a3d4e61c828"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[]}
