)]}'
{"id":"openstack%2Fvenus~851637","triplet_id":"openstack%2Fvenus~master~I4a15c86a4a18bfb44bba632c62730173da1f010c","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-08-02 01:13:18.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-08-11 09:00:10.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-08-11 08:49:43.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I4a15c86a4a18bfb44bba632c62730173da1f010c","subject":"Supplement test cases for logs.","status":"MERGED","created":"2022-08-01 12:08:57.000000000","updated":"2022-08-11 09:01:05.000000000","submitted":"2022-08-11 09:00:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851637","meta_rev_id":"78012015c922e609ab7d6dd7ee9974f155829b63","_number":851637,"virtual_id_number":851637,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 09:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-08-11 00:50:03.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-08-11 08:49:43.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-08-11 08:49:43.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-01 12:14:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-01 12:25:26.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-08-01 12:25:26.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"648dbcd96838bea46b337757adac6dbd5a804c93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-01 12:08:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b8734c052ec62cbf2bf10980f35c65ea7d666c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 12:14:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9adb7b906ea4b3781282e128d7ac80a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c196dc515434264a2873c01cdac5f4f : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab0e58388be48108ba1752b035c3e15 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24c8bf33a2bc4e178cd88a1041983940 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5ff155c291a68cdd03ecea0d8838d11bbe68db2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-01 12:23:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad9ff79681b9eb87656b06fd59aa3f6eb0243d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 12:33:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd027b0cc0a44018d1be245b062cc35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d0c304fd634318aa3c634afff261cf : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985c55bad3c94caabba7e150d33c6a8c : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cb42ff9dc734d2d81caf8cf1286ffb9 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3c0b12702b6de423ed3cae00ed7a478dca2cd033","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-02 01:13:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ff2946bf7b7b4e5b694c75e8e585a72df52d3be0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-10 09:47:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f95798c60fd0fdad0f8eb5ccc6c43373b0ec627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:59:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a50bf3637234410b973823d54986f9cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cbb357f7b2740e887a4504b8e798c5e : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1da7bf3ff8b41e08634cc72b9646ff6 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48de47e260d041e9b92df3717f0a990d : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":3},{"id":"02084e51382ae176bbbd51481262ab1c4c590e9f","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-11 00:50:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"440b99065bf0f5c893da69fa303bee01f832df6c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-11 08:49:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"24e881663e903886e80bd9666ce59545ec7722b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 08:50:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c4f6c99ebbaf22c8c84c7c90b0c59b02cc33346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 09:00:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7311fd19e4a74c41af7ccc908d595784\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787522338b834dde9a530182850a65fa : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6b07438ac946e4b166a8a90f91e34a : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc3b3990accd4317b84509a91386edd5 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":3},{"id":"231ca81e25704da9fc69f07579384ae162e2f210","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 09:00:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"78012015c922e609ab7d6dd7ee9974f155829b63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 09:01:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab1f2de3f7d4399a28167c27b7a51a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07eff62e92aa4af8ba7d860d2900128c : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2953ee4b1744ca9c5130eb90d5e2c3e161cd95c7","revisions":{"3f5b6dd204424f326b439575d1107d159b452468":{"kind":"REWORK","_number":1,"created":"2022-08-01 12:08:57.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/37/851637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/37/851637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/37/851637/1"}}},"commit":{"parents":[{"commit":"d098d4a0d70a2415d3164c07c44fc66c917e2610","subject":"Another test case to supplement logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d098d4a0d70a2415d3164c07c44fc66c917e2610"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:05:27.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:05:27.000000000","tz":480},"subject":"Supplement test cases for logs.","message":"Supplement test cases for logs.\n\nIn test_logs_no_buckets，the case without buckets is added.\n\nChange-Id: I4a15c86a4a18bfb44bba632c62730173da1f010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3f5b6dd204424f326b439575d1107d159b452468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3f5b6dd204424f326b439575d1107d159b452468"}]},"branch":"refs/heads/master"},"653bb9f126d1979b9500926c11774e99f1d5519b":{"kind":"REWORK","_number":2,"created":"2022-08-01 12:23:52.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/37/851637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/37/851637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/37/851637/2"}}},"commit":{"parents":[{"commit":"d098d4a0d70a2415d3164c07c44fc66c917e2610","subject":"Another test case to supplement logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d098d4a0d70a2415d3164c07c44fc66c917e2610"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:05:27.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:23:33.000000000","tz":480},"subject":"Supplement test cases for logs.","message":"Supplement test cases for logs.\n\nIn test_logs_no_buckets，the case without buckets is added.\n\nChange-Id: I4a15c86a4a18bfb44bba632c62730173da1f010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/653bb9f126d1979b9500926c11774e99f1d5519b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/653bb9f126d1979b9500926c11774e99f1d5519b"}]},"branch":"refs/heads/master"},"2953ee4b1744ca9c5130eb90d5e2c3e161cd95c7":{"kind":"REWORK","_number":3,"created":"2022-08-10 09:47:43.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/37/851637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/37/851637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/851637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/37/851637/3"}}},"commit":{"parents":[{"commit":"ca570a8793bd15f2b976323de8883a3d4e61c828","subject":"Add a test cases for \u0027logs\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ca570a8793bd15f2b976323de8883a3d4e61c828"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:05:27.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-10 09:47:14.000000000","tz":480},"subject":"Supplement test cases for logs.","message":"Supplement test cases for logs.\n\nIn test_logs_no_buckets，the case without buckets is added.\n\nChange-Id: I4a15c86a4a18bfb44bba632c62730173da1f010c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2953ee4b1744ca9c5130eb90d5e2c3e161cd95c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2953ee4b1744ca9c5130eb90d5e2c3e161cd95c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[]}
