)]}'
{"id":"openstack%2Fvenus~851638","triplet_id":"openstack%2Fvenus~master~Ic25572e0f78f7551ef588f0ab4212b98733c853f","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-08-02 01:13:40.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-08-10 09:21:11.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-08-04 07:38:23.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"Ic25572e0f78f7551ef588f0ab4212b98733c853f","subject":"Add a test case of logs to the test file.","status":"MERGED","created":"2022-08-01 12:15:34.000000000","updated":"2022-08-10 09:22:11.000000000","submitted":"2022-08-10 09:21:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851638","meta_rev_id":"5a15e5524aef9ec5e07282cb165e716d29e0b40c","_number":851638,"virtual_id_number":851638,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 09:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-08-10 09:14:58.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-08-04 07:38:23.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2022-08-10 09:14:58.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-01 12:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-01 12:25:52.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-08-01 12:25:52.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"3e46394d887079caa66f8ce318588521496868b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-01 12:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9ebe2f0f1654646e9cd3a2f35beb013f72719e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 12:25:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312e3d93902940fb911f527f7a9a1b54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1182cb0f664138882e7077feb776a6 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f31677afd942abbd879676dd87ac72 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bfdb87506664939a6db6a66b7844ea8 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d7dd79b94b5fef7a63b13f89e055c86112763652","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-02 01:13:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e1a20874fd1178fd9412075fa91d34c74162f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-08-04 06:50:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd8bfa0a6ed4809589968acf799d53066d4f8007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 06:56:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9744579e0e954432a0a0c9da169e64ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feecaf4abe4f4b66ac9025b284510589 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84f3e79eac784204a560c536186a7f8f : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc9f0f7e80bf496cb4beda8df8f7d76d : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c890c7b7a304ade28911a804539c20fdf05902ad","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-04 07:38:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2602f0840a69c7a1e933046bb57ab67cb9047970","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-10 09:14:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"17813477d40ac90c0784d86e1988f3ffdd5c73f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:15:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"009a330878582309a7288f11fa80426941caa903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:21:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46983c7eadf14e8a900ad80a06934caf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331bacb7326a449ea6465725ff77f751 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b675a9df4b43cab917c3d4471023fc : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d3a265c97ef44e18a00af8aeb60a44f : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"68c26115d2c11b1ef7ac85ae35d1c0b46c6e7ae6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:21:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5a15e5524aef9ec5e07282cb165e716d29e0b40c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:22:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6945601174b4f719a30ab6e03cf486f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ae8ed4c1434de0a07977731d232721 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b9ab46be3a25c35f1ec035b61cde713149f86ff","revisions":{"e4f2b26e58bfe5410f5bfb4c06ddef0ccd0a5626":{"kind":"REWORK","_number":1,"created":"2022-08-01 12:15:34.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/38/851638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/38/851638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/38/851638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/38/851638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/38/851638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/38/851638/1"}}},"commit":{"parents":[{"commit":"d098d4a0d70a2415d3164c07c44fc66c917e2610","subject":"Another test case to supplement logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d098d4a0d70a2415d3164c07c44fc66c917e2610"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:13:07.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:13:07.000000000","tz":480},"subject":"Add a test case of logs to the test file.","message":"Add a test case of logs to the test file.\n\nA test case for logs: test_logs_no_hit, which prints\n\"no data, no hit\".\n\nChange-Id: Ic25572e0f78f7551ef588f0ab4212b98733c853f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e4f2b26e58bfe5410f5bfb4c06ddef0ccd0a5626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e4f2b26e58bfe5410f5bfb4c06ddef0ccd0a5626"}]},"branch":"refs/heads/master"},"5b9ab46be3a25c35f1ec035b61cde713149f86ff":{"kind":"REWORK","_number":2,"created":"2022-08-04 06:50:22.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/38/851638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/38/851638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/38/851638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/38/851638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/38/851638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/38/851638/2"}}},"commit":{"parents":[{"commit":"f305b4f9e8101913ec23cebb4f8a9f5cca871faa","subject":"Add a use case for the test unit.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f305b4f9e8101913ec23cebb4f8a9f5cca871faa"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-01 12:13:07.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-08-04 06:49:56.000000000","tz":480},"subject":"Add a test case of logs to the test file.","message":"Add a test case of logs to the test file.\n\nA test case for logs: test_logs_no_hit, which prints\n\"no data, no hit\".\n\nChange-Id: Ic25572e0f78f7551ef588f0ab4212b98733c853f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5b9ab46be3a25c35f1ec035b61cde713149f86ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5b9ab46be3a25c35f1ec035b61cde713149f86ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[]}
