)]}'
{"id":"openstack%2Fvenus~854955","triplet_id":"openstack%2Fvenus~master~Iad6aea8d4d8d3c5b488396edc7bdd716c37881a5","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-08-30 00:53:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad6aea8d4d8d3c5b488396edc7bdd716c37881a5","subject":"Rename venus to openstack-venus","status":"MERGED","created":"2022-08-29 09:11:43.000000000","updated":"2022-08-30 00:54:38.000000000","submitted":"2022-08-30 00:53:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854955","meta_rev_id":"ba3a60289d17506aeb74bcb55d433f78740f95b0","_number":854955,"virtual_id_number":854955,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-30 00:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-08-30 00:48:13.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-29 11:23:29.000000000","_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2022-08-30 00:48:13.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 09:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 09:17:56.000000000","updated_by":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"reviewer":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"state":"REVIEWER"},{"updated":"2022-08-30 00:48:13.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"1d38f98b335b00eaa934cb9e130d5b055b1166b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-29 09:11:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"757030c1a74470704b8e20b283222f973f1af5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 09:16:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ad6f47c2289454f8feba747eb9a952e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c02e1b96952f4f3dacc530f72a95b2ce : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d45dcf95b8c1432680d5a90190d5588b : FAILURE in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21fc29eeee154dae84bc124c2f8905d9 : FAILURE in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7e96a05359e4b98a81c42ca19649a2b : FAILURE in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"942bacd6f275d3f77429c90dba20408c05e31339","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-29 09:17:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db08f3d32259f9719abf1aa90e58b1cf6a804411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-29 09:45:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f58cc75bae225a5feaaf82d7dbb905f5eeb37ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 09:55:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e521ddee037d44499e9ea939b6987463\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a9b9ed72868472bace56c2d6e85cf57 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07b64601756450fa15dcc1ed0fc2c6d : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b93793fdaf54368a6747ab24bdac405 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c16f9728db74426cb3048848b7c930ca : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7ea65ebd3cd63b9d7c58cec872c007d3862db79f","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-29 11:23:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6b4b2167679229ba1c0383eb92d77a018edfb219","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-30 00:48:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99e60cb7491d14db74db8db619cfea5b9647bc4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 00:48:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16ffdf58bd4ebc19fbc90fe9a85fd9b2aefc5504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 00:53:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a656c93b11d4185a099267b3859e473\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5767aed70a57447ba49552660c92ef0d : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ab670e34b54dd2a329c1549743ca62 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c7166a7416347f992724d0b725d4fef : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02e19457e2d04aac90605e4ead749e42 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b1a72a78d254b383b87bbd1d163ce43a1ea8c843","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 00:53:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ba3a60289d17506aeb74bcb55d433f78740f95b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 00:54:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1984ba7a60d84f55ae58d5fecb8817bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375f92ae8d8249fdab2d22ed9b64a7d8 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f060b59cede279d722b916de422bab453c1411dd","revisions":{"25c668c6c6bfebe5d19c91815a105459de8d64ab":{"kind":"REWORK","_number":1,"created":"2022-08-29 09:11:43.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/55/854955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/55/854955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/55/854955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/55/854955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/55/854955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/55/854955/1"}}},"commit":{"parents":[{"commit":"1a6d7e5591811426eb35ac716c8a4e1a69c7b324","subject":"Add a test case of typical_stats to the test file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1a6d7e5591811426eb35ac716c8a4e1a69c7b324"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 09:08:54.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 09:11:30.000000000","tz":480},"subject":"Rename venus to openstack-venus","message":"Rename venus to openstack-venus\n\nWe cannot bulid venus directory in PYPI because it has created [1], so we\nrename to openstack-venus to create directory in PYPI.\n\n[1]https://zuul.opendev.org/t/openstack/build/3e690921d50644b89c34b375b9faf4b7\n\nChange-Id: Iad6aea8d4d8d3c5b488396edc7bdd716c37881a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/25c668c6c6bfebe5d19c91815a105459de8d64ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/25c668c6c6bfebe5d19c91815a105459de8d64ab"}]},"branch":"refs/heads/master"},"f060b59cede279d722b916de422bab453c1411dd":{"kind":"REWORK","_number":2,"created":"2022-08-29 09:45:38.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/55/854955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/55/854955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/55/854955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/55/854955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/55/854955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/55/854955/2"}}},"commit":{"parents":[{"commit":"1a6d7e5591811426eb35ac716c8a4e1a69c7b324","subject":"Add a test case of typical_stats to the test file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1a6d7e5591811426eb35ac716c8a4e1a69c7b324"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 09:08:54.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 09:45:09.000000000","tz":480},"subject":"Rename venus to openstack-venus","message":"Rename venus to openstack-venus\n\nWe cannot bulid venus directory in PYPI because it has created [1], so we\nrename to openstack-venus to create directory in PYPI.\n\n[1]https://zuul.opendev.org/t/openstack/build/3e690921d50644b89c34b375b9faf4b7\n\nChange-Id: Iad6aea8d4d8d3c5b488396edc7bdd716c37881a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f060b59cede279d722b916de422bab453c1411dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f060b59cede279d722b916de422bab453c1411dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[]}
