)]}'
{"id":"openstack%2Fvenus~856388","triplet_id":"openstack%2Fvenus~master~I6dec42a9687eeab005f07904cee621593cace969","project":"openstack/venus","branch":"master","attention_set":{"34340":{"account":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"last_update":"2022-09-11 02:47:39.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6dec42a9687eeab005f07904cee621593cace969","subject":"add @mock.path() to test_params_no_index_data","status":"NEW","created":"2022-09-11 02:34:32.000000000","updated":"2022-09-11 02:47:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2909a4d31577a9aa8b47facfaf32125d84e491d5","_number":856388,"virtual_id_number":856388,"owner":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-11 02:47:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 02:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65fca5130ec48ede08411cf73f832157f55afe24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"date":"2022-09-11 02:34:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2909a4d31577a9aa8b47facfaf32125d84e491d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 02:47:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed243229eac247989df92db79b2fc24f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b502a3bdbc4393aa7dbf51df80dc36 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f2333336c44e47a83369b6d0550836 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff279b4e084441afb9b5227e076fe9c0 : FAILURE in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3da7ae8242fb4eb8b35c0b03c951ff9d : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/348b4b18c62c4dc6949bf41f27cc3848 : FAILURE in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d4e06873327c3e4c7b1a63d165cb63a58ede1ff","revisions":{"8d4e06873327c3e4c7b1a63d165cb63a58ede1ff":{"kind":"REWORK","_number":1,"created":"2022-09-11 02:34:32.000000000","uploader":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"ref":"refs/changes/88/856388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/88/856388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/88/856388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/88/856388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/88/856388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/88/856388/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-09-11 02:33:09.000000000","tz":480},"committer":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-09-11 02:33:09.000000000","tz":480},"subject":"add @mock.path() to test_params_no_index_data","message":"add @mock.path() to test_params_no_index_data\n\nadd @mock.path() to test_params_no_index_data method\n\nChange-Id: I6dec42a9687eeab005f07904cee621593cace969\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8d4e06873327c3e4c7b1a63d165cb63a58ede1ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8d4e06873327c3e4c7b1a63d165cb63a58ede1ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
