)]}'
{"id":"openstack%2Fvenus~856389","triplet_id":"openstack%2Fvenus~master~I4a63b7a06d3a861520bd0c748b9ba5ac03c5fb41","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2022-09-11 08:24:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4a63b7a06d3a861520bd0c748b9ba5ac03c5fb41","subject":"add api doc: the custom config","status":"ABANDONED","created":"2022-09-11 06:01:02.000000000","updated":"2022-09-11 08:24:52.000000000","total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"2e66540901db31f87f276bfb60934ea3112a1082","_number":856389,"virtual_id_number":856389,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-11 06:09:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":-1,"date":"2022-09-11 08:01:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 06:09:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-11 08:01:29.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"bc9f975db9cdb807bc6827ec9dbd223306c2d7f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-09-11 06:01:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d90c4ae69a3fbc1d0922bcb65343b0e10792b792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 06:09:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/136e18916d2840c2b633e3a4b61be87d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2eef0106e36442f2b82c28e6369b614f : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59616c83e7f441ff916aa1735111ec4e : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c7c9c3e088348a7b00b9f0592fe5415 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"15831e28669a3e6c34ea3ff1031effd9b870f90d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-11 08:01:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2e66540901db31f87f276bfb60934ea3112a1082","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-09-11 08:24:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97e706f336484e06e4a0815ffac24c625393a5c3","revisions":{"97e706f336484e06e4a0815ffac24c625393a5c3":{"kind":"REWORK","_number":1,"created":"2022-09-11 06:01:02.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/89/856389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/89/856389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/89/856389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/89/856389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/89/856389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/89/856389/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 06:00:17.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 06:00:40.000000000","tz":480},"subject":"add api doc: the custom config","message":"add api doc: the custom config\n\nChange-Id: I4a63b7a06d3a861520bd0c748b9ba5ac03c5fb41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/97e706f336484e06e4a0815ffac24c625393a5c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/97e706f336484e06e4a0815ffac24c625393a5c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
