)]}'
{"id":"openstack%2Fvenus~856605","triplet_id":"openstack%2Fvenus~master~If03b294d682c384d0d6482fa64cc1df2e6b05052","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2022-09-09 09:08:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If03b294d682c384d0d6482fa64cc1df2e6b05052","subject":"depandance the module collect logs","status":"MERGED","created":"2022-09-09 08:23:44.000000000","updated":"2022-09-09 09:09:52.000000000","submitted":"2022-09-09 09:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856605","meta_rev_id":"d9630833f203c8275bf2fb9f88cd9d02331d1e8c","_number":856605,"virtual_id_number":856605,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 09:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2022-09-09 08:58:21.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2022-09-09 08:58:21.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 08:29:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 08:58:21.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"4c076bdecc94acce1f2aa7ce6a7a9dc152b251e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-09-09 08:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dea2e070f4579cd010dc285d5624b3b661cf300c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 08:29:23.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f25c65c65d54a268aceb4a11b897d6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e38868f8290485f8a50de1857af4626 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21cbffe452774563815583169226b18c : FAILURE in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a7828889fda4b3c9e4fdafce08e1658 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7db9143cf5ea45e98ddff27495cb813b : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa223f5308394c4ea9ff34e1f6152058 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1533b122eef02e040bea350b8c1847f394a41556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-09-09 08:36:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f37c9a959eff1c84f5bcb48238bec4316f5210b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 08:48:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b854de00d64167a3551544444edbf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4edd40a1714da9a993b25b71d5c9ed : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69dff2a618274e5c9f7df8486c3ca275 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5fa1dd637b747e5b0b11b3e0196736a : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fc30b2c3caf4d2690514a034f3e297d : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c86d11d0a74485694ed213fcd4b5663 : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11212ee29c20ab77c2fa1c2a0780ec483ea42391","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2022-09-09 08:58:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9261d1fe28be5087bda9eb53c59c355933298fdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 08:58:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b87b07b7c6de8f4a84671e3678023fecc2b88be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 09:08:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9937a0e0fe45424c8d9705d4258d3e03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1275dc785be41c4a66b6df254f13e74 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d42db12736964493ae95331f24a6a650 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bda41997152e4e53a78942c9b8017748 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4fa9549c6e24483993178c4a95bdcfc : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"fff3599625f918f7cb7e61cae8f9295da280f338","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 09:08:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d9630833f203c8275bf2fb9f88cd9d02331d1e8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 09:09:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81767680a1e441788d8a3c9b829edbcf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e748bc490e6f48b8ba95c6c677db8bc6 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b52728e4fb0224ccaa475d4e665d6c2041d0e98","revisions":{"b3c00ccb3b251d19b913e1046eb74882a92bb547":{"kind":"REWORK","_number":1,"created":"2022-09-09 08:23:44.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/05/856605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/05/856605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/05/856605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/05/856605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/05/856605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/05/856605/1"}}},"commit":{"parents":[{"commit":"13cf26a03fc4b82789e6dbb57a15a1f22adb5839","subject":"add missing lib of project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/13cf26a03fc4b82789e6dbb57a15a1f22adb5839"}]}],"author":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2022-09-09 08:23:29.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2022-09-09 08:23:29.000000000","tz":480},"subject":"depandance the module collect logs","message":"depandance the module collect logs\n\nChange-Id: If03b294d682c384d0d6482fa64cc1df2e6b05052\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b3c00ccb3b251d19b913e1046eb74882a92bb547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b3c00ccb3b251d19b913e1046eb74882a92bb547"}]},"branch":"refs/heads/master"},"9b52728e4fb0224ccaa475d4e665d6c2041d0e98":{"kind":"REWORK","_number":2,"created":"2022-09-09 08:36:39.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/05/856605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/05/856605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/05/856605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/05/856605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/05/856605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/05/856605/2"}}},"commit":{"parents":[{"commit":"13cf26a03fc4b82789e6dbb57a15a1f22adb5839","subject":"add missing lib of project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/13cf26a03fc4b82789e6dbb57a15a1f22adb5839"}]}],"author":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2022-09-09 08:23:29.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui3@163.com","date":"2022-09-09 08:36:26.000000000","tz":480},"subject":"depandance the module collect logs","message":"depandance the module collect logs\n\nChange-Id: If03b294d682c384d0d6482fa64cc1df2e6b05052\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/9b52728e4fb0224ccaa475d4e665d6c2041d0e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/9b52728e4fb0224ccaa475d4e665d6c2041d0e98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}}]}],"submit_requirements":[]}
