)]}'
{"id":"openstack%2Fvenus~856930","triplet_id":"openstack%2Fvenus~master~I7a980942697c40d115b65663cfdfdda2ee4691e5","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34340":{"account":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"last_update":"2022-09-14 01:21:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a980942697c40d115b65663cfdfdda2ee4691e5","subject":"add @mock.patch to test_get_all_index empty","status":"MERGED","created":"2022-09-11 06:19:16.000000000","updated":"2022-09-14 01:22:40.000000000","submitted":"2022-09-14 01:21:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856930","meta_rev_id":"5342490a23a2107b32020134255b2af4c3a93ab6","_number":856930,"virtual_id_number":856930,"owner":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 01:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2022-09-14 01:08:04.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-13 08:01:10.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2022-09-14 01:08:04.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 06:29:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-13 08:01:10.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-09-14 01:08:04.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"5addb97c31997c422bd6f766252f03603afd004b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"date":"2022-09-11 06:19:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90629e8df72908bf1772654596e8327f04aa07f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 06:29:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dcef6f006354be985ae0ae4badee1bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc573204b4d84fab869b530ed1e3d6cd : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eee5d679da143d0930efb873db8e6da : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50bd57a92d4d4eb9943a24d65a9224a9 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7e9ef22be34e6b96aa797753362d29 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b126af6eaca04c1eb462a174941111d6 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57df42a4e9595c207abdd1f77bfcb72d0aa3afb0","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-13 08:01:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c64d0dbde14afb984073493b8f59bfe4baf7df73","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-09-14 01:08:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e484572cf6af95298df501b4e86aaaf39837c11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:08:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46f75884771878d69f928cfea1efaaf91aa1fc50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:21:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d50415235e34d78ac105653a1db1df0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2edde3dbfe05451d815520eeab426c35 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366f0721dcf54c5d8fef5961bf0b7843 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b964274fb0684fffb079a575e2603709 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/533f797b0c824ac3a1e0b48e1b1cf9df : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"94160c8b2da784550f331f5babea13574b5347c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:21:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5342490a23a2107b32020134255b2af4c3a93ab6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:22:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202f81788733485b9c246f0a7259284c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4dd7e5b527f4123808d181f12720626 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af9aa76e28530ee9ef5af43544748a992ad6fd8a","revisions":{"af9aa76e28530ee9ef5af43544748a992ad6fd8a":{"kind":"REWORK","_number":1,"created":"2022-09-11 06:19:16.000000000","uploader":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"ref":"refs/changes/30/856930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/30/856930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/856930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/856930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/856930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/30/856930/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-09-11 06:18:14.000000000","tz":480},"committer":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-09-11 06:18:14.000000000","tz":480},"subject":"add @mock.patch to test_get_all_index empty","message":"add @mock.patch to test_get_all_index empty\n\nadd @mock.patch to test_get_all_index_empty method\n\nChange-Id: I7a980942697c40d115b65663cfdfdda2ee4691e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/af9aa76e28530ee9ef5af43544748a992ad6fd8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/af9aa76e28530ee9ef5af43544748a992ad6fd8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
