)]}'
{"id":"openstack%2Fvenus~856932","triplet_id":"openstack%2Fvenus~master~Id201272f675085d2774e6bc390aaa2529159c4e4","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34340":{"account":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"last_update":"2022-09-14 01:23:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id201272f675085d2774e6bc390aaa2529159c4e4","subject":"add @mock.patch to test_get_all_index","status":"MERGED","created":"2022-09-11 06:28:49.000000000","updated":"2022-09-14 01:24:09.000000000","submitted":"2022-09-14 01:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856932","meta_rev_id":"c4e918891ace885a870b52628c36c5b4a387a96a","_number":856932,"virtual_id_number":856932,"owner":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 01:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2022-09-14 01:08:21.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-13 08:00:53.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2022-09-14 01:08:21.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 06:37:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-13 08:00:53.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-09-14 01:08:21.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"faca80b8fd73f8d61a42db3159d27b4ef2a0699e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"date":"2022-09-11 06:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bded740d58d36e7af2651c2534f09ce7c6723f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 06:37:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3c67b305c949779b73d7f1c1e5e7f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ae42714f67427cb6ee5ff956c347f0 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59405deb875d4933a3387d6886e87ef0 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1e865db8ca2476facd8b05f1b9938bf : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e79512554c4969b02aaf8713332e59 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66aa6d0d98bd4e33b99a7a32fc7aaa8d : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b47aee86ffb03868b642326c2ad95b0809e9874","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-13 08:00:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"524dfd6ba228f3ba3b28da26f82bd4ca42d0c12d","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-09-14 01:08:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d498fe769b2ed7a93ce4f949e9a0aaa200130de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:08:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a31efed4601fcef0818fd71fc3432c3eed10cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:23:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce06eb4e84214eb3839c66e7c211ddda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52108efe36d54c7fab11e4831d9141e9 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641250ef2696437ba23dc5b964c6939a : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08d4ae9b08a644688fa78141087d6631 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a5d4efc163b40bbab7f137680cfc85a : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f85e37f0ba638256670b0637d87b0435ac65c0d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:23:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c4e918891ace885a870b52628c36c5b4a387a96a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 01:24:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f26d2640d8c41caad6a7046901a5947\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578ea07f546f476a8d77ae11b96b9052 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e41c086e7c93c0350cbd94794b1a3912e96e2935","revisions":{"e41c086e7c93c0350cbd94794b1a3912e96e2935":{"kind":"REWORK","_number":1,"created":"2022-09-11 06:28:49.000000000","uploader":{"_account_id":34340,"name":"shanyunfan33","display_name":"shanyunfan","email":"shanyunfan@inspur.com","username":"shanyunfan","status":"working"},"ref":"refs/changes/32/856932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/32/856932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/32/856932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/32/856932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/32/856932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/32/856932/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-09-11 06:27:45.000000000","tz":480},"committer":{"name":"shanyunfan33","email":"shanyunfan@inspur.com","date":"2022-09-11 06:27:45.000000000","tz":480},"subject":"add @mock.patch to test_get_all_index","message":"add @mock.patch to test_get_all_index\n\nadd @mock.patch to test_get_all_index method\n\nChange-Id: Id201272f675085d2774e6bc390aaa2529159c4e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e41c086e7c93c0350cbd94794b1a3912e96e2935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e41c086e7c93c0350cbd94794b1a3912e96e2935"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
