)]}'
{"id":"openstack%2Fvenus~856934","triplet_id":"openstack%2Fvenus~master~I1a84121c2013e7142f87f6f6ac1eae0bac62f1b4","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32020":{"account":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"last_update":"2022-09-14 10:30:28.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-09-14 10:30:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a84121c2013e7142f87f6f6ac1eae0bac62f1b4","subject":"Switch to 2023.1 Python3 unit tests and generic template name","status":"MERGED","created":"2022-09-11 06:42:01.000000000","updated":"2022-09-15 08:29:53.000000000","submitted":"2022-09-14 10:30:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856934","meta_rev_id":"04526d67b8fa8b1ed0a9c7f27d9c64ff3f74b862","_number":856934,"virtual_id_number":856934,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 10:30:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":2,"date":"2022-09-14 10:16:07.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-14 10:17:10.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-14 10:17:10.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 06:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-11 06:59:30.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-14 10:17:10.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-09-15 08:29:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"e4bf611814c64d2ba514558085aa5ad5fbfc6647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-11 06:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"131fc04757c94573bcc97e366c00a676fb75fa6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 06:52:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/772ffe5f713141c6b65122f1df9d4cce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a507be2e4e4ff3b7378d4165de1310 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db441a7fa762445f8be08ca3f2411efe : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5c1c54696c4cef8d3911a80fa1c1a4 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44f9032b526a4d1ea8ea55d766570cf1 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5e49f2c9f3f4d9924210acce0f0ced8ee7d3cce7","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-11 06:59:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"442bc9fc56746fd538c6d67bec22f2c5af0da586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-14 10:14:27.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"59138832ea069e5575f23d37ae67381b0376c52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-14 10:15:13.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ce94a93f97fbbf5d9db216aa254f3c22c1363b0f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-14 10:16:07.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ff1ae5291b0b92eebe2b47b72680926070fcc38","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-14 10:17:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1bbd64914acc264b71a50ea79a416bd1154bdf5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 10:21:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4b027e80ff49c5a8402c2f44f016e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076b708e16b146479d5af655e9c32828 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63e685637c14bd8a2e496aa3c6c1f7c : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/becf55673a9d4c25b696482765f2c4cc : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56a5724349b14624ba8ef1bc14e8f1fc : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39d34cb3eefc41e8b341c1813cdedcc1 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ec60dc778578e78b97fdd670cb5f77fc5ace034d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 10:22:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"497a5c9ab673b5d8bdfea7a61a7a673127c282b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 10:30:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e18469880516421794b9943c03c24104\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a32dd5ae7834a51b33a670b5ecb7800 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef10033e57348c69447a35f844dfb94 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/510219c9b1db463d8612cac0a5218f95 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57383a5dd6424f41b2251b6ed920b50f : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"70b0a2a6fa25eecd0f4b85ce83a151eee425185f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 10:30:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b3439205dc5250fbfc2cb1b3f8f28943fd519baa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 10:32:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc3499bb7d9e4cd5ba68272f2e4299f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43449261c9d7402e9a818a3073e831cc : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3},{"id":"04526d67b8fa8b1ed0a9c7f27d9c64ff3f74b862","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-15 08:29:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6a797aaa6b4bee7c5c799c481e6919f8e4b44254","revisions":{"37fb84446b2383659daed3d02ed496141bf7e53c":{"kind":"REWORK","_number":1,"created":"2022-09-11 06:42:01.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/34/856934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/34/856934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/34/856934/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-11 06:16:11.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-11 06:41:35.000000000","tz":480},"subject":"Add Python3 antelope unit tests","message":"Add Python3 antelope unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I1a84121c2013e7142f87f6f6ac1eae0bac62f1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/37fb84446b2383659daed3d02ed496141bf7e53c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/37fb84446b2383659daed3d02ed496141bf7e53c"}]},"branch":"refs/heads/master"},"d59eaf641b3fc6eed7c0728071c43fcc7ed86c0b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-14 10:14:27.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/34/856934/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/34/856934/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/34/856934/2"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-11 06:16:11.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2022-09-14 10:14:27.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\nChange-Id: I1a84121c2013e7142f87f6f6ac1eae0bac62f1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d59eaf641b3fc6eed7c0728071c43fcc7ed86c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d59eaf641b3fc6eed7c0728071c43fcc7ed86c0b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6a797aaa6b4bee7c5c799c481e6919f8e4b44254":{"kind":"REWORK","_number":3,"created":"2022-09-14 10:15:13.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/34/856934/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/34/856934/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/34/856934/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/34/856934/3"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-11 06:16:11.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2022-09-14 10:15:13.000000000","tz":0},"subject":"Switch to 2023.1 Python3 unit tests and generic template name","message":"Switch to 2023.1 Python3 unit tests and generic template name\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for antelope. Also,\nupdating the template name to generic one.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\nChange-Id: I1a84121c2013e7142f87f6f6ac1eae0bac62f1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6a797aaa6b4bee7c5c799c481e6919f8e4b44254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6a797aaa6b4bee7c5c799c481e6919f8e4b44254"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
