)]}'
{"id":"openstack%2Fvenus~856937","triplet_id":"openstack%2Fvenus~master~I2e143da9ee83de98d6dd6f530239c6ee09593ee9","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2022-09-12 01:07:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e143da9ee83de98d6dd6f530239c6ee09593ee9","subject":"Add api doc: the search logs","status":"MERGED","created":"2022-09-11 07:14:42.000000000","updated":"2022-09-12 01:07:54.000000000","submitted":"2022-09-12 01:07:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"856937","meta_rev_id":"06df89096593baf1e4005c20bb6596fcc4a0c878","_number":856937,"virtual_id_number":856937,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 01:07:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 00:55:20.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 00:55:20.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 07:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-11 07:50:27.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-12 00:55:20.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"ad773426c0f8631ba758c457446e2c712145db51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-09-11 07:14:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3af0db74016d3e6faa6757f0528342a61847d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 07:21:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fffa7519c304c2da01ea4031d84ccda\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9781311e959040fab06d3f8f8ddfff89 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cbfc03dd524609b4569cc080fdb396 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03af8d2dc21842e38c55868e16f96cb1 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6fe9f973090456d13f55d0ac517c5aede1e42ef5","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-11 07:50:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81a8907182ba5ce9025841a0777ea21f76ab9e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-09-11 08:22:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"52fff1b3dbd6c65d1f7c88d63db7f3380f3bcdd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 08:37:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7912087d63d46c597af1fdf413cb4d2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/10ae5bca4af34f999e667e0016b7fcf5 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072b292e77f74beda4c9100e24b6f504 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbd64ae8b224e1cb617cfcac5e7f9eb : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"26dee1b6ad5995e0837b114bb22f50b9ad557f3f","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 00:55:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a65668dfab9665f27d51063b2d55fbf69db4d44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 00:55:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"708df8a58fcb379df25bc37e108c11d14ffdee06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:07:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7421edd34464683a25dd642374a1df2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f3156540124640a0a0d954f123cf57bc : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/947553c3e7d44fa584d59141eba57b66 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4276f6f2744d5a858d17acde0b92e8 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"20039da86fcd23becf5be1b0571a90fcf28267f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:07:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"06df89096593baf1e4005c20bb6596fcc4a0c878","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:07:54.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18733665f30244ddbb4671556e8a0d40\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/da3ce66bbfac4fd79ddc426bdcaec341 : FAILURE in 36s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3d95f339da4f6a8aeee57fd33c40fe : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0429ad8cdf08596bcddd6c23099dd51c5f28656b","revisions":{"d2108cd2aebcc3229b42f0d2d0801a7c2533cd49":{"kind":"REWORK","_number":1,"created":"2022-09-11 07:14:42.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/37/856937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/37/856937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/856937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/856937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/856937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/37/856937/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 07:14:02.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 07:14:18.000000000","tz":480},"subject":"Add api doc: the search logs","message":"Add api doc: the search logs\n\nChange-Id: I2e143da9ee83de98d6dd6f530239c6ee09593ee9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d2108cd2aebcc3229b42f0d2d0801a7c2533cd49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d2108cd2aebcc3229b42f0d2d0801a7c2533cd49"}]},"branch":"refs/heads/master"},"0429ad8cdf08596bcddd6c23099dd51c5f28656b":{"kind":"REWORK","_number":2,"created":"2022-09-11 08:22:54.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/37/856937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/37/856937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/856937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/856937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/856937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/37/856937/2"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 07:14:02.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 08:22:20.000000000","tz":480},"subject":"Add api doc: the search logs","message":"Add api doc: the search logs\n\nChange-Id: I2e143da9ee83de98d6dd6f530239c6ee09593ee9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/0429ad8cdf08596bcddd6c23099dd51c5f28656b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/0429ad8cdf08596bcddd6c23099dd51c5f28656b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
