)]}'
{"id":"openstack%2Fvenus~856942","triplet_id":"openstack%2Fvenus~master~I6929a6d1a9c6bfd0bda1a66b8b1ca89ffd87e470","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2022-09-12 01:07:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6929a6d1a9c6bfd0bda1a66b8b1ca89ffd87e470","subject":"add api doc: the search instance request ids","status":"MERGED","created":"2022-09-11 07:35:59.000000000","updated":"2022-09-12 01:08:45.000000000","submitted":"2022-09-12 01:07:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"856942","meta_rev_id":"9c38f6d4cdc7b4c78e92fb088a74c85a9e7e6cb4","_number":856942,"virtual_id_number":856942,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 01:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 00:56:26.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 00:56:26.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 07:46:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-11 07:58:35.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-12 00:56:26.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"01b971d983bca6c4d8a44e311e71fca8b3e601d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-09-11 07:35:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0d8d82c788e87e1bf072bfe9f208f518e0d3e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 07:46:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd3825b06af4f3ea16114cc67bb1130\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c3d67b1dc37c42dfa386250cdb4ccefc : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6efd2359c89a46e28439d93dea5915d7 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838e48c5b6564e78bede2bbbcb05b1a0 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0ff91a668e040fe23c2a5b7f16fc1798fb2a1804","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-11 07:58:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e5204703c784a987f4bf7a4ce25f2c7b010b3dba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-09-11 08:15:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ee4372d57fb0dbdc671d9aea2bf1100e6eadd90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 08:21:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a555c48b241a47f79135f68c1ce8ebcb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/37e0a6356f074891a70f2713cac6de57 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9129e7d9966400cbb98ba994bc7b8d0 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5032b2fbddfb44c6a15d7679d4afe2d2 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6187ae2e21fea78a88dc197672c24766db3607f6","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 00:56:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92a126c8e432b319ccea727086b6f35c1c6c3cb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 00:57:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9620c0f83cd6e23159dc10528da5fd25543ba39f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:07:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd54273921bd499ea03468190f78eff5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/24c6411d5f5b40eca561539532896274 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f2c667e979454084122daa99a31613 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1486fb090f448048e39684c7c3fe75c : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"aac38cac14ccb2ae6601eeb00fa9bec76cebe35a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:07:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9c38f6d4cdc7b4c78e92fb088a74c85a9e7e6cb4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:08:45.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/037ae7fafbf245f683cd52e81fd0752d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/881c24f2b3f643ae8902f095f88bb735 : FAILURE in 34s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d869dd7c0d34cdc87ade98deaf4cb16 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"70382b3d5ed87b0dc56a70944d39546f650930a3","revisions":{"0c4ce9bf1a9e6d188c7f5449e4d3524f232c58ab":{"kind":"REWORK","_number":1,"created":"2022-09-11 07:35:59.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/42/856942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/42/856942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/42/856942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/42/856942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/42/856942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/42/856942/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 07:35:18.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 07:35:35.000000000","tz":480},"subject":"add api doc: the search instance request ids","message":"add api doc: the search instance request ids\n\nChange-Id: I6929a6d1a9c6bfd0bda1a66b8b1ca89ffd87e470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/0c4ce9bf1a9e6d188c7f5449e4d3524f232c58ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/0c4ce9bf1a9e6d188c7f5449e4d3524f232c58ab"}]},"branch":"refs/heads/master"},"70382b3d5ed87b0dc56a70944d39546f650930a3":{"kind":"REWORK","_number":2,"created":"2022-09-11 08:15:30.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/42/856942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/42/856942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/42/856942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/42/856942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/42/856942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/42/856942/2"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 07:35:18.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 08:14:57.000000000","tz":480},"subject":"add api doc: the search instance request ids","message":"add api doc: the search instance request ids\n\nChange-Id: I6929a6d1a9c6bfd0bda1a66b8b1ca89ffd87e470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/70382b3d5ed87b0dc56a70944d39546f650930a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/70382b3d5ed87b0dc56a70944d39546f650930a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
