)]}'
{"id":"openstack%2Fvenus~856946","triplet_id":"openstack%2Fvenus~master~Ic7c7fae095f7be2cbc4f4bc5d6b513f2e9fa268f","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2022-09-12 01:05:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7c7fae095f7be2cbc4f4bc5d6b513f2e9fa268f","subject":"add api doc: the response of get custom config","status":"MERGED","created":"2022-09-11 08:28:08.000000000","updated":"2022-09-12 01:06:18.000000000","submitted":"2022-09-12 01:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856946","meta_rev_id":"e6429204b5a2ec4239404d5c628ef21ddad1592f","_number":856946,"virtual_id_number":856946,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 01:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 00:54:57.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 00:54:57.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-11 08:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 00:54:57.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"e855de3a45e64ee10b6fdbd4bafa4f97a40a5f66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-09-11 08:28:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f3798e60323b0398344c726226227c1f6192b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 08:43:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52c4f604530141b284b2f37bb07c1f30\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c427d6ab410b4ee08c1fb21d2d57df3e : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ff85e6b4004619873dc701663268a9 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abac7f1325494c49adc62ec33768d9ea : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"018d3cc7bb28e847ddf7ce1cdf56a93eaecd8ee7","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 00:54:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f9b2caca680a122acd132bad4f0f148cd486ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 00:55:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22f920b7bea02fc7d9139c27aa40e268e9f71782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:05:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/577a215ccff146d7bf0eaccd6e66f799\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6c7e721b2f92458693e207c0be07e9c1 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4475794175df40f88a6c067309c47f64 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6efc9212934bd29ac3df9f8923de7d : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1077b243b25f49f8d085306d1e8362a294d82210","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:05:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6429204b5a2ec4239404d5c628ef21ddad1592f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:06:18.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2089efcc2bef402e9aae9c99ce23eac0\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e8996ecc166b4c1a9c2bc2161073a0bc : FAILURE in 33s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f8c9a1ad7241d4b528668aeabbe20f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df7ae3ff4c6777b463564534eea3645661782fdf","revisions":{"df7ae3ff4c6777b463564534eea3645661782fdf":{"kind":"REWORK","_number":1,"created":"2022-09-11 08:28:08.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/46/856946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/46/856946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/46/856946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/46/856946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/46/856946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/46/856946/1"}}},"commit":{"parents":[{"commit":"bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1","subject":"Merge \"depandance the module collect logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bd1b4b7a2f322c65f3b1e1cb48df8d0c70a3e9b1"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 08:27:28.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-09-11 08:27:45.000000000","tz":480},"subject":"add api doc: the response of get custom config","message":"add api doc: the response of get custom config\n\nChange-Id: Ic7c7fae095f7be2cbc4f4bc5d6b513f2e9fa268f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/df7ae3ff4c6777b463564534eea3645661782fdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/df7ae3ff4c6777b463564534eea3645661782fdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
