)]}'
{"id":"openstack%2Fvenus~856956","triplet_id":"openstack%2Fvenus~master~Ifd8c3e3f44bb345a84fcd07a9edfe470a52ead17","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32020":{"account":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"last_update":"2022-09-12 06:45:19.000000000","reason":"Change was abandoned"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-09-12 06:45:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifd8c3e3f44bb345a84fcd07a9edfe470a52ead17","subject":"Add api doc: the search logs releate request","status":"ABANDONED","created":"2022-09-12 02:57:16.000000000","updated":"2022-09-12 06:45:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"73f8fbe23931a60d219511460f0fbd9a0ec5bd50","_number":856956,"virtual_id_number":856956,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-12 06:34:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 02:58:02.000000000","updated_by":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-12 03:03:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d99a8cdae88c7032beb9b94bbfeac449e3c5ab6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 02:57:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69e2a2f493be558bc124ea81cb59103b0cff462e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 03:03:33.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cec55da5e9fb44549145787523c8d6c0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/12e17d30c28c44399f7adf68a9c3d0cd : FAILURE in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84adf0668d454b648e043465f5efae59 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e8c233efcfa4adea4a596c23095b41b : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"519f696d95533dc5794a8fb6b175cea9d78c8690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 05:52:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"09802b7434a4e1e5d06cb8ddf79ffe59f63cd83b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:00:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4bd3c612824537b0c12442de6b5bc3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9260b52b9ddb43f2a8848ea6331f3c34 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2b19d146a1434fb08626df093fef07 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496952660733457784b04e20ebe4dff0 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"73efcb2c6bbe347510e8112510797fe2c3ca11c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 06:23:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"31e82de03a3735570599090cd1c73b791ade21e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:34:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36fe3b3812474d45b75454dc57484c29\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/370de35051534c78ad931417fedfe606 : FAILURE in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce6a2c00a4249fca69992cb00e29369 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c55f53848043778a94c00dcdd06a64 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":3},{"id":"73f8fbe23931a60d219511460f0fbd9a0ec5bd50","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 06:45:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"45f1e855671646a7c32d8117daa531eacc6f0378","revisions":{"b50a0895a0b6988d6f4eb80b2800f7898fa8aaaf":{"kind":"REWORK","_number":1,"created":"2022-09-12 02:57:16.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/56/856956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/56/856956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/56/856956/1"}}},"commit":{"parents":[{"commit":"fa48fa5bdbd3ab72df4293d59d52af4dbf3a33d2","subject":"use patameter yaml to record","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/fa48fa5bdbd3ab72df4293d59d52af4dbf3a33d2"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 02:56:03.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 02:56:03.000000000","tz":480},"subject":"Add api doc: the search logs releate request","message":"Add api doc: the search logs releate request\n\nChange-Id: Ifd8c3e3f44bb345a84fcd07a9edfe470a52ead17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b50a0895a0b6988d6f4eb80b2800f7898fa8aaaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b50a0895a0b6988d6f4eb80b2800f7898fa8aaaf"}]},"branch":"refs/heads/master"},"676dfde501844c61c02bd5e13bfa7a68e14332e1":{"kind":"REWORK","_number":2,"created":"2022-09-12 05:52:04.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/56/856956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/56/856956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/56/856956/2"}}},"commit":{"parents":[{"commit":"fa48fa5bdbd3ab72df4293d59d52af4dbf3a33d2","subject":"use patameter yaml to record","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/fa48fa5bdbd3ab72df4293d59d52af4dbf3a33d2"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 02:56:03.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 05:50:03.000000000","tz":480},"subject":"Add api doc: the search logs releate request","message":"Add api doc: the search logs releate request\n\nChange-Id: Ifd8c3e3f44bb345a84fcd07a9edfe470a52ead17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/676dfde501844c61c02bd5e13bfa7a68e14332e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/676dfde501844c61c02bd5e13bfa7a68e14332e1"}]},"branch":"refs/heads/master"},"45f1e855671646a7c32d8117daa531eacc6f0378":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-12 06:23:56.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/56/856956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/56/856956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/56/856956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/56/856956/3"}}},"commit":{"parents":[{"commit":"006a0dec64e4627291b4d519d61316abd0827204","subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/006a0dec64e4627291b4d519d61316abd0827204"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 02:56:03.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 06:23:56.000000000","tz":0},"subject":"Add api doc: the search logs releate request","message":"Add api doc: the search logs releate request\n\nChange-Id: Ifd8c3e3f44bb345a84fcd07a9edfe470a52ead17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/45f1e855671646a7c32d8117daa531eacc6f0378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/45f1e855671646a7c32d8117daa531eacc6f0378"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
