)]}'
{"id":"openstack%2Fvenus~856957","triplet_id":"openstack%2Fvenus~master~I504fd2ce7fc9c5130391ba088a51c3bbcc355380","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32020":{"account":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"last_update":"2022-09-12 06:45:38.000000000","reason":"Change was abandoned"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-09-12 06:45:38.000000000","reason":"Change was abandoned"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-09-12 06:45:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I504fd2ce7fc9c5130391ba088a51c3bbcc355380","subject":"Add api doc: the search logs download releate request","status":"ABANDONED","created":"2022-09-12 03:09:45.000000000","updated":"2022-09-12 06:45:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f22542a776aec923bdff18095400d5ec92a1c20","_number":856957,"virtual_id_number":856957,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-12 06:35:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 03:15:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 03:16:22.000000000","updated_by":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-12 03:17:16.000000000","updated_by":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"d885fe7cd6d2667021c1ab2fff6420f987f122a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 03:09:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5d3d0d2282f3c83b793f445b5b8502d02594209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 03:15:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1721cb2e53d47ec9de5169b8c164621\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7f1402f1866b410abf1566cd85227859 : FAILURE in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7309643b57a64e58a19fe42580f794e0 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8fd8717b0c4916aafae30d83d66e37 : SUCCESS in 4m 25s\n\nWarning:\n  Comments left for invalid file api-ref/source/v1/search-logs.inc","accounts_in_message":[],"_revision_number":1},{"id":"c5dcad7ed54d08e561ec064fa5e7f71620806e63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 05:58:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a09923abf4756cf1bdf04f80f2b0fcce9be977a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 05:59:30.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/venus for 856956,2","accounts_in_message":[],"_revision_number":2},{"id":"2a6a3d138734e1f7a232b9cb929797deb7d269ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 06:27:17.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"956aa72e629f470fb03dfd42db7ba8344d90ab6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:35:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfb79dbc02a749de9c29ba4ede577985\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de7e23da85d64fe5b2a8b6b5801f1a2f : FAILURE in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465d8b87c5974b8e96272488285b01ca : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f3917fc09744dbb32042d07df665b1 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6f22542a776aec923bdff18095400d5ec92a1c20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 06:45:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7d66138637741cec614f9fe8e30270c51cf69a8b","revisions":{"8612befe0f533943057c5c8ce80675e6e6e58911":{"kind":"REWORK","_number":1,"created":"2022-09-12 03:09:45.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/57/856957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/57/856957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/57/856957/1"}}},"commit":{"parents":[{"commit":"b50a0895a0b6988d6f4eb80b2800f7898fa8aaaf","subject":"Add api doc: the search logs releate request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b50a0895a0b6988d6f4eb80b2800f7898fa8aaaf"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 03:08:38.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 03:08:38.000000000","tz":480},"subject":"Add api doc: the search logs download releate request","message":"Add api doc: the search logs download releate request\n\nChange-Id: I504fd2ce7fc9c5130391ba088a51c3bbcc355380\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8612befe0f533943057c5c8ce80675e6e6e58911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8612befe0f533943057c5c8ce80675e6e6e58911"}]},"branch":"refs/heads/master"},"d4b43ec15c7466f3be060fa0460243ec689f5c2c":{"kind":"REWORK","_number":2,"created":"2022-09-12 05:58:58.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/57/856957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/57/856957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/57/856957/2"}}},"commit":{"parents":[{"commit":"b50a0895a0b6988d6f4eb80b2800f7898fa8aaaf","subject":"Add api doc: the search logs releate request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b50a0895a0b6988d6f4eb80b2800f7898fa8aaaf"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 03:08:38.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 05:57:26.000000000","tz":480},"subject":"Add api doc: the search logs download releate request","message":"Add api doc: the search logs download releate request\n\nChange-Id: I504fd2ce7fc9c5130391ba088a51c3bbcc355380\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d4b43ec15c7466f3be060fa0460243ec689f5c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d4b43ec15c7466f3be060fa0460243ec689f5c2c"}]},"branch":"refs/heads/master"},"7d66138637741cec614f9fe8e30270c51cf69a8b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-12 06:27:17.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/57/856957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/57/856957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/57/856957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/57/856957/3"}}},"commit":{"parents":[{"commit":"45f1e855671646a7c32d8117daa531eacc6f0378","subject":"Add api doc: the search logs releate request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/45f1e855671646a7c32d8117daa531eacc6f0378"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 03:08:38.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 06:27:17.000000000","tz":0},"subject":"Add api doc: the search logs download releate request","message":"Add api doc: the search logs download releate request\n\nChange-Id: I504fd2ce7fc9c5130391ba088a51c3bbcc355380\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7d66138637741cec614f9fe8e30270c51cf69a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7d66138637741cec614f9fe8e30270c51cf69a8b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
