)]}'
{"id":"openstack%2Fvenus~856961","triplet_id":"openstack%2Fvenus~master~Iafd99b33899dc3540e78465021f711eae0abe08b","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2022-09-12 06:02:24.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-09-12 06:02:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafd99b33899dc3540e78465021f711eae0abe08b","subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","status":"MERGED","created":"2022-09-12 04:38:23.000000000","updated":"2022-09-12 06:03:26.000000000","submitted":"2022-09-12 06:02:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856961","meta_rev_id":"c819b9e6f0072f6593681631f9848ac473fdde20","_number":856961,"virtual_id_number":856961,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 06:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2022-09-12 05:54:47.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2022-09-12 05:54:47.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 04:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 04:57:24.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2022-09-12 05:54:47.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"28e07d816543a2925ce3113a030625397ee6ac34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 04:38:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c98d03d6e0a5a6cac2f28673cd30d6c8ccb1b966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 04:43:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5310fb5fd7734346b404785f2e396280\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bc685ef0b8f64cbd96b54afb9eed3f4a : FAILURE in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f668de5944b54cfaab8e00014b61cc81 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a3d2be9beb430488b656d426a65838 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"62a48064ddbb52450e510528744fb2520259c068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 04:53:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b35329c3b8bce65a577acbbfd1856259635645e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 05:01:25.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef6aa44b8814418384ccdad94ee5b38f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/63acd49babdb46e09930e1ea375c06d0 : FAILURE in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753c4dc0de784ddb8fd6259c18c20555 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf3d40690514ce3a3f646827f325fec : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"114ae2c7bcb5b564c1e627c114195fac31e32603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 05:41:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"216f80e120565f9e90fd570de76ab5137eba26d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 05:49:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09316823de4a44ebb9cdf47b7259b11a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aa02eb6b48104285b138b985c0168b22 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca30502f01e1484b8acd670a2272b9db : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76be0bab0d7945218835837d0f039b4b : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b2b09f02e57024b5fd9ebd4fe0657b6a337ef165","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2022-09-12 05:54:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2a731a8dbc441adad0815eaeb99129fac1711648","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 05:55:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"157c3cce8a62b7757390674096821c0d4abab805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:02:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2cc9250ccbd44509587f2d9c3cbfa71\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/54e494432dbf4bc49bc131bb239ffc4d : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e855d247a7604431af93f39a50f0a891 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec1e05597aa45f9afd7723dfeaaff56 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":3},{"id":"e26df4ab2d6934479a9ec83a6e8c4e8f8d551598","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:02:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c819b9e6f0072f6593681631f9848ac473fdde20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:03:26.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e1c40a1b9d24e47ab970af85ba54585\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/01228c86185943efaccf2ce98c15b588 : FAILURE in 35s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e4be006e4a4b9a9d342376e1f2bc27 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"006a0dec64e4627291b4d519d61316abd0827204","revisions":{"a25719e088605b4aa8541aeb44c7bddb57d252aa":{"kind":"REWORK","_number":1,"created":"2022-09-12 04:38:23.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/61/856961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/61/856961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/61/856961/1"}}},"commit":{"parents":[{"commit":"40adfdd5f12b52d351e06515f747d43723f0c668","subject":"Merge \"use patameter yaml to record\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/40adfdd5f12b52d351e06515f747d43723f0c668"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 04:37:57.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 04:38:09.000000000","tz":0},"subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","message":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc\n\nChange-Id: Iafd99b33899dc3540e78465021f711eae0abe08b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a25719e088605b4aa8541aeb44c7bddb57d252aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a25719e088605b4aa8541aeb44c7bddb57d252aa"}]},"branch":"refs/heads/master"},"52698edd5d12ebc33028fe6cc11ebc4d911ee81c":{"kind":"REWORK","_number":2,"created":"2022-09-12 04:53:47.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/61/856961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/61/856961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/61/856961/2"}}},"commit":{"parents":[{"commit":"40adfdd5f12b52d351e06515f747d43723f0c668","subject":"Merge \"use patameter yaml to record\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/40adfdd5f12b52d351e06515f747d43723f0c668"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 04:37:57.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 04:53:10.000000000","tz":0},"subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","message":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc\n\nChange-Id: Iafd99b33899dc3540e78465021f711eae0abe08b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/52698edd5d12ebc33028fe6cc11ebc4d911ee81c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/52698edd5d12ebc33028fe6cc11ebc4d911ee81c"}]},"branch":"refs/heads/master"},"006a0dec64e4627291b4d519d61316abd0827204":{"kind":"REWORK","_number":3,"created":"2022-09-12 05:41:21.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/61/856961/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/61/856961/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/61/856961/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/61/856961/3"}}},"commit":{"parents":[{"commit":"40adfdd5f12b52d351e06515f747d43723f0c668","subject":"Merge \"use patameter yaml to record\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/40adfdd5f12b52d351e06515f747d43723f0c668"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 04:37:57.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 05:41:00.000000000","tz":0},"subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","message":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc\n\nChange-Id: Iafd99b33899dc3540e78465021f711eae0abe08b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/006a0dec64e4627291b4d519d61316abd0827204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/006a0dec64e4627291b4d519d61316abd0827204"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}}]}],"submit_requirements":[]}
