)]}'
{"id":"openstack%2Fvenus~856963","triplet_id":"openstack%2Fvenus~master~Ia0d71643891936d71931e385946a846ac5b5b5dc","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-09-12 06:59:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0d71643891936d71931e385946a846ac5b5b5dc","subject":"change custom-config.inc file name and add some notes","status":"MERGED","created":"2022-09-12 06:27:07.000000000","updated":"2022-09-12 07:00:32.000000000","submitted":"2022-09-12 06:59:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856963","meta_rev_id":"1e1ef36ed1584d7d334166c2cefe39e435576b49","_number":856963,"virtual_id_number":856963,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 06:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2022-09-12 06:51:20.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2022-09-12 06:51:20.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 06:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 06:51:20.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"332a5d69db052708a4a88a08115aa5c085d2fdea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 06:27:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69a518079cf1deb9d7d7b5f71da399d694004c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:37:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/290323a741e24cbf9522ef26b5299554\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d1000c8563ee49a5a7e7fff7707f0767 : FAILURE in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f47c8a092a402f97fd9959cb3f7d59 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6df5e91a6f34774aaaf8c9e7862d7c1 : SUCCESS in 5m 50s\n\nWarning:\n  Comments left for invalid file api-ref/source/v1/index.rst","accounts_in_message":[],"_revision_number":1},{"id":"2d48cf1406b20321dc4f698ea96cd32bf6954b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 06:41:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"658968e1ea3ec7178dfeea80d0a78f543c119130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:47:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5b5869e7e6b4c028b2540f47c85f2f2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/20e7b5875dae41769e60b4ded5502d41 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99335e79bce3483194aa186d18148544 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea1f1adbb374a618d12351b1354fd01 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0086803e2bb5cbb283ad1344d9246e363986cedc","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2022-09-12 06:51:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35b9cd3afadcd1dc28c8be5e9306598c037dfab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:51:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7adacb0202a7195831319066d120828518e6d056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:59:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/660d34c28ea04db4b10b1148dab59bca\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2044ff88e83c4862a4b44cd105ebf456 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa76e4e770e74eed97a3dd85605f04e2 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498673a98e1240dcb254d466023b8be0 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c76b9de508c82f905b5c7dbf7b46e84cd39ca709","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:59:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1e1ef36ed1584d7d334166c2cefe39e435576b49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:00:32.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e20ed10f2ca643a89e992951b352e7d9\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/37fe01f2d1fe4f389ad77e2705407ec8 : FAILURE in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b12c6fe79fb4e7a9b7e3ddd030a523f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f54142ab32ced8523e1fc366666358f36ad283b6","revisions":{"be777048399a0d9e937d7f80e2d9221c1001f379":{"kind":"REWORK","_number":1,"created":"2022-09-12 06:27:07.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/63/856963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/63/856963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/856963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/856963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/856963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/63/856963/1"}}},"commit":{"parents":[{"commit":"006a0dec64e4627291b4d519d61316abd0827204","subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/006a0dec64e4627291b4d519d61316abd0827204"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 06:26:48.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 06:26:48.000000000","tz":0},"subject":"change custom-config.inc file name and add some notes","message":"change custom-config.inc file name and add some notes\n\nChange-Id: Ia0d71643891936d71931e385946a846ac5b5b5dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/be777048399a0d9e937d7f80e2d9221c1001f379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/be777048399a0d9e937d7f80e2d9221c1001f379"}]},"branch":"refs/heads/master"},"f54142ab32ced8523e1fc366666358f36ad283b6":{"kind":"REWORK","_number":2,"created":"2022-09-12 06:41:03.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/63/856963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/63/856963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/856963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/856963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/63/856963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/63/856963/2"}}},"commit":{"parents":[{"commit":"006a0dec64e4627291b4d519d61316abd0827204","subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/006a0dec64e4627291b4d519d61316abd0827204"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 06:26:48.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 06:40:43.000000000","tz":0},"subject":"change custom-config.inc file name and add some notes","message":"change custom-config.inc file name and add some notes\n\nChange-Id: Ia0d71643891936d71931e385946a846ac5b5b5dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f54142ab32ced8523e1fc366666358f36ad283b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f54142ab32ced8523e1fc366666358f36ad283b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}}]}],"submit_requirements":[]}
