)]}'
{"id":"openstack%2Fvenus~856965","triplet_id":"openstack%2Fvenus~master~I3e067493429d7b443cd35ed71d0aae395ae5870e","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32020":{"account":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"last_update":"2022-09-12 07:10:46.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-09-12 07:10:46.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-09-12 07:02:56.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}},"hashtags":[],"change_id":"I3e067493429d7b443cd35ed71d0aae395ae5870e","subject":"Add api doc: the search download logs releate request","status":"MERGED","created":"2022-09-12 06:52:02.000000000","updated":"2022-09-12 07:12:28.000000000","submitted":"2022-09-12 07:10:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856965","meta_rev_id":"bf6124a891e42162a416dcf378bd8e2819cf8e86","_number":856965,"virtual_id_number":856965,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 07:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 07:02:56.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 07:02:56.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 06:55:53.000000000","updated_by":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-09-12 06:55:53.000000000","updated_by":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-12 06:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39804e40536fc97f9a57fbe31ae8f6151e732d4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 06:52:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ceff668ce75e73e980ba5aba5f4535008b229ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:57:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0a41985aa44f19ac83847933c07184\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/512c245d8dc04e9993d11a479d47346a : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9e64f5a91740b1ab65811122dc2648 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addc892440ed4f8a95470ce43eb86453 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"fdc93ef626996d9297028453408a20cfe5d7f52b","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 07:02:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a694052229a0139efa7a531fa3063759d0252c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:03:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c6bba7ab97abe4c391adc74f188c117da5aa19a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:10:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff4b6c3b48fa408ea8280adf3cc43901\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bed08598a69642aa81fd50557b2c20e2 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae98891fec34a2bb8568d340e5f1155 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efe125c211343758732d9c5abb25331 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e7663bd61a99059218e200e1120561db6641ac52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:10:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf6124a891e42162a416dcf378bd8e2819cf8e86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:12:28.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acf97301638b4774b6559aa81b889dc0\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/9642ca1cef3340d4ad79379906df85ec : FAILURE in 46s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d7d3f92684e442fb86f927e3c394419 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"734e5eed5b842e07109255dd08e27a575fa24de4","revisions":{"734e5eed5b842e07109255dd08e27a575fa24de4":{"kind":"REWORK","_number":1,"created":"2022-09-12 06:52:02.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/65/856965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/65/856965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/856965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/856965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/856965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/65/856965/1"}}},"commit":{"parents":[{"commit":"006a0dec64e4627291b4d519d61316abd0827204","subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/006a0dec64e4627291b4d519d61316abd0827204"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 06:50:48.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 06:50:48.000000000","tz":480},"subject":"Add api doc: the search download logs releate request","message":"Add api doc: the search download logs releate request\n\nChange-Id: I3e067493429d7b443cd35ed71d0aae395ae5870e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/734e5eed5b842e07109255dd08e27a575fa24de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/734e5eed5b842e07109255dd08e27a575fa24de4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
