)]}'
{"id":"openstack%2Fvenus~856966","triplet_id":"openstack%2Fvenus~master~I74423c2b45032361116e1d2965456f3fab5f2b13","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32020":{"account":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"last_update":"2022-09-12 07:10:12.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-09-12 07:10:12.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-09-12 07:02:18.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}},"hashtags":[],"change_id":"I74423c2b45032361116e1d2965456f3fab5f2b13","subject":"Add api doc: the search analyse logs releate request","status":"MERGED","created":"2022-09-12 06:54:38.000000000","updated":"2022-09-12 07:11:31.000000000","submitted":"2022-09-12 07:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856966","meta_rev_id":"34e84956f00dda8c533518d59106204f06291f14","_number":856966,"virtual_id_number":856966,"owner":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-12 07:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 07:02:18.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 07:02:18.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 06:55:35.000000000","updated_by":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-09-12 06:55:35.000000000","updated_by":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-12 07:00:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a46ddda9098744cd35b4b60a2cbe87a6c241896","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"date":"2022-09-12 06:54:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c4af6bae4108db052e2c0e0fbde0cdb3364b0c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:00:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/553f9c71137a4381a7985d44d7ba0547\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8c35c27794644b4bb5e456d3c6666c1d : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af79fdf7b7148958b7a253e75e8a3e5 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492dea2f10364740bd56febe146facf0 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c6e12d7f09a3ae17d91dc8947212d8fc2700fc4b","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 07:02:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0710f0aaa0516f9e487122e27d0a4ecaa833684d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:02:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe506b150c5d531045a2ebd41fbac4fad6ec5702","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:10:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/974cc46a46c74ce19248d6582734124d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8719f81e331a45e3b1dbb6638771a616 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11db13650574eec879e75c8cb67d53e : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd5f1d2fca346358e1706bddb0d9f04 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"181b6d06a361c9f3254c504321385024161bdc33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:10:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"34e84956f00dda8c533518d59106204f06291f14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:11:31.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/281a16f617334ec6901864e362518141\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/5f55bb5c05b24c1180399fc101835f73 : FAILURE in 36s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e29bfad360b4fb7874a380b459f10e3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88d2851e52811019daf872b4b4d2bd8c4aae8b7c","revisions":{"88d2851e52811019daf872b4b4d2bd8c4aae8b7c":{"kind":"REWORK","_number":1,"created":"2022-09-12 06:54:38.000000000","uploader":{"_account_id":32020,"name":"zhangtongjian","email":"zhangtj03@inspur.com","username":"zhangtongjian"},"ref":"refs/changes/66/856966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/66/856966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/856966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/856966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/856966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/66/856966/1"}}},"commit":{"parents":[{"commit":"006a0dec64e4627291b4d519d61316abd0827204","subject":"recover status.yaml \u0026 touch empty v1/parameters.yaml \u0026 add custom config api doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/006a0dec64e4627291b4d519d61316abd0827204"}]}],"author":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 06:53:30.000000000","tz":480},"committer":{"name":"zhangtongjian","email":"125163227@qq.com","date":"2022-09-12 06:53:30.000000000","tz":480},"subject":"Add api doc: the search analyse logs releate request","message":"Add api doc: the search analyse logs releate request\n\nChange-Id: I74423c2b45032361116e1d2965456f3fab5f2b13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/88d2851e52811019daf872b4b4d2bd8c4aae8b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/88d2851e52811019daf872b4b4d2bd8c4aae8b7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
