)]}'
{"id":"openstack%2Fvenus~857011","triplet_id":"openstack%2Fvenus~master~I75032bec21a5ffb7e4e7ca37fbd3c446fed10ea3","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-09-12 08:17:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I75032bec21a5ffb7e4e7ca37fbd3c446fed10ea3","subject":"add search log inc files in v1/index.rst","status":"ABANDONED","created":"2022-09-12 07:17:21.000000000","updated":"2022-09-12 08:17:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f1d6564c32437c5a4eb9a3bd4a7a77b42574931","_number":857011,"virtual_id_number":857011,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-12 08:14:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 07:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a058b578f7a13d872fa77cf69a5fb263c08f1d13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 07:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"020e3b983cce78a1851fdadbcd5d340c452031f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:27:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73e15f257c4a4441aa9acc9883c946b9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c1320dfcee934a45b05c499801b11a12 : FAILURE in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac300b952374a3db5eb871f7a4ae778 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4df3102c17946099dad7796f1737634 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"73a3ac0222e94011f194da1b3a898f63f9c552bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 08:03:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"773ef03a11a2eb48147ddac20bb7e9d185700c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 08:14:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/615e1b9d41c84362b41c6ac6fe9ccad1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4ea50c36949e44af84dc446ef305feaf : FAILURE in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7543d5e35644afb0b9ee9e6c0d8499 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a53d6d41734689b3bb0d5c059997d7 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"6f1d6564c32437c5a4eb9a3bd4a7a77b42574931","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-12 08:17:14.000000000","message":"Abandoned\n\naffected other files","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"48cdad16a3691683a1de523b124d5330cf8cf05b","revisions":{"1962ff62f02010a42e56083777e09232fff3da1c":{"kind":"REWORK","_number":1,"created":"2022-09-12 07:17:21.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/11/857011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/11/857011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/11/857011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/11/857011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/11/857011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/11/857011/1"}}},"commit":{"parents":[{"commit":"4e206fedece3fc7cd179b45a36854534977a020a","subject":"Merge \"Add api doc: the search download logs releate request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4e206fedece3fc7cd179b45a36854534977a020a"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 07:17:04.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 07:17:04.000000000","tz":0},"subject":"add search log inc files in v1/index.rst","message":"add search log inc files in v1/index.rst\n\nChange-Id: I75032bec21a5ffb7e4e7ca37fbd3c446fed10ea3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1962ff62f02010a42e56083777e09232fff3da1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1962ff62f02010a42e56083777e09232fff3da1c"}]},"branch":"refs/heads/master"},"48cdad16a3691683a1de523b124d5330cf8cf05b":{"kind":"REWORK","_number":2,"created":"2022-09-12 08:03:43.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/11/857011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/11/857011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/11/857011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/11/857011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/11/857011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/11/857011/2"}}},"commit":{"parents":[{"commit":"4e206fedece3fc7cd179b45a36854534977a020a","subject":"Merge \"Add api doc: the search download logs releate request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4e206fedece3fc7cd179b45a36854534977a020a"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 07:17:04.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-12 08:03:43.000000000","tz":0},"subject":"add search log inc files in v1/index.rst","message":"add search log inc files in v1/index.rst\n\nChange-Id: I75032bec21a5ffb7e4e7ca37fbd3c446fed10ea3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/48cdad16a3691683a1de523b124d5330cf8cf05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/48cdad16a3691683a1de523b124d5330cf8cf05b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
