)]}'
{"id":"openstack%2Fvenus~857306","triplet_id":"openstack%2Fvenus~master~Iba1ed8ab01bb0a148e78beb388bccb3cacd93f1c","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-09-14 06:24:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iba1ed8ab01bb0a148e78beb388bccb3cacd93f1c","subject":"sort api in index.rst \u0026 fix status code \u0026 add response json of two apis","status":"MERGED","created":"2022-09-14 04:09:23.000000000","updated":"2022-09-14 06:25:42.000000000","submitted":"2022-09-14 06:24:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"857306","meta_rev_id":"11039aad498968918585e13f3856e050237385f7","_number":857306,"virtual_id_number":857306,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 06:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2022-09-14 06:18:15.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-14 06:18:17.000000000","_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2022-09-14 06:18:15.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-14 06:18:17.000000000","_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 04:17:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-14 05:41:47.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-14 06:18:15.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2022-09-14 06:18:17.000000000","updated_by":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"reviewer":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"state":"REVIEWER"}],"messages":[{"id":"a23123f5a7120d5d706456ef077f981427a83ecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-14 04:09:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efde50658699d2521bba12b33f733248eba736f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 04:17:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48209d8a9ee4220a8d9a97541562b47\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e692feb25fa34d33a44d173c3c9e54ec : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50d9c4f45594a15aca12565ff871d62 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8d3d248b5e40b2a73317c4c3ce5c8a : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a78a613a09f8dcf3fe1f4cf53e435a068ccc37fb","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-14 05:41:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6b5c6e122e2a878adf2998b169ce706373c4051b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-14 05:49:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61759282ba6174a36b4ecad5286c07bd6d8a5cc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-09-14 05:53:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0663cdc00e3837d716751168ea6fec89b882096b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 06:00:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96856f96fa9a4bf8a9c61bca2f8ed1a0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44842dcb391945ea9f20a47b36de0a5e : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9eac004bc44388aad6b656d32f4e00 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce2996482c14b8caa67900217f85bbf : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"bdbee01dd75ca4ddf80be9221139e98f5d14abf2","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2022-09-14 06:18:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9152385f23c6a27f5c95e136fe28bc94a44f624c","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-09-14 06:18:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f18749ab8e52712690abe24da99b29f1e267a3cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 06:18:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2041cf798eaea925f072f4d213e3d225229fe623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 06:24:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/843160cd093347028241d022b988a050\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/640b107f3ee447379b702550bc9f5662 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9eaeba76224581950d60b677433b13 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aec2d4deb3a469da3c28c958df94ba0 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1a39bd6a128b8b85d018ee8a7d500bd357d22606","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 06:24:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"11039aad498968918585e13f3856e050237385f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 06:25:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e5934a52a1c4d528ac9f1b2f6dbe811\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ce377a52ef764a43b208e7847b343ad6 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889736b555d9485ebf25b88af0f9ca04 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3be9f638f767d786db29e57c71f198d21b2bd8f3","revisions":{"ef56187dcb9fa9d8093fb7e492681742262df014":{"kind":"REWORK","_number":1,"created":"2022-09-14 04:09:23.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/06/857306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/06/857306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/06/857306/1"}}},"commit":{"parents":[{"commit":"3720f31587c4bd18cdc6f82e3728d50f2526cd41","subject":"Merge \"add log index and fix title underline short error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3720f31587c4bd18cdc6f82e3728d50f2526cd41"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-14 04:09:06.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-14 04:09:06.000000000","tz":0},"subject":"sort api in index.rst \u0026 fix status code \u0026 add response json of two apis","message":"sort api in index.rst \u0026 fix status code \u0026 add response json of two apis\n\nChange-Id: Iba1ed8ab01bb0a148e78beb388bccb3cacd93f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ef56187dcb9fa9d8093fb7e492681742262df014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ef56187dcb9fa9d8093fb7e492681742262df014"}]},"branch":"refs/heads/master"},"ce3e72e487f458e82655c87b518927bbcdee5adf":{"kind":"REWORK","_number":2,"created":"2022-09-14 05:49:34.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/06/857306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/06/857306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/06/857306/2"}}},"commit":{"parents":[{"commit":"3720f31587c4bd18cdc6f82e3728d50f2526cd41","subject":"Merge \"add log index and fix title underline short error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3720f31587c4bd18cdc6f82e3728d50f2526cd41"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-14 04:09:06.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-14 05:48:10.000000000","tz":0},"subject":"sort api in index.rst \u0026 fix status code \u0026 add response json of two apis","message":"sort api in index.rst \u0026 fix status code \u0026 add response json of two apis\n\nChange-Id: Iba1ed8ab01bb0a148e78beb388bccb3cacd93f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ce3e72e487f458e82655c87b518927bbcdee5adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ce3e72e487f458e82655c87b518927bbcdee5adf"}]},"branch":"refs/heads/master"},"3be9f638f767d786db29e57c71f198d21b2bd8f3":{"kind":"REWORK","_number":3,"created":"2022-09-14 05:53:27.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/06/857306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/06/857306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/857306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/06/857306/3"}}},"commit":{"parents":[{"commit":"3720f31587c4bd18cdc6f82e3728d50f2526cd41","subject":"Merge \"add log index and fix title underline short error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3720f31587c4bd18cdc6f82e3728d50f2526cd41"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-14 04:09:06.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2022-09-14 05:53:11.000000000","tz":0},"subject":"sort api in index.rst \u0026 fix status code \u0026 add response json of two apis","message":"sort api in index.rst \u0026 fix status code \u0026 add response json of two apis\n\nChange-Id: Iba1ed8ab01bb0a148e78beb388bccb3cacd93f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3be9f638f767d786db29e57c71f198d21b2bd8f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3be9f638f767d786db29e57c71f198d21b2bd8f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}}]}],"submit_requirements":[]}
