)]}'
{"id":"openstack%2Fvenus~862792","triplet_id":"openstack%2Fvenus~master~I141e0b249af692a4c46a80577251a6a33cb54c6a","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-10-28 02:40:59.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-10-28 02:40:59.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-10-28 02:15:25.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}},"hashtags":[],"change_id":"I141e0b249af692a4c46a80577251a6a33cb54c6a","subject":"Add default conf for es","status":"MERGED","created":"2022-10-27 10:21:11.000000000","updated":"2022-10-28 02:42:25.000000000","submitted":"2022-10-28 02:40:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"862792","meta_rev_id":"4bac1b845564b75765e1fc5a36b0538a3b939b81","_number":862792,"virtual_id_number":862792,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 02:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-28 02:23:17.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-28 02:25:25.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 10:26:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-28 01:55:20.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-10-28 01:55:20.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"fb442140c9e9a1ee893c20c41f66439cd1c147a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-10-27 10:21:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07a52f9ef8cb0a831ed9a8943503ed1a4d3ac025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:26:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9297efb64234d0b973fd4fa5ffaa4ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa46b78fed84245a8b715007f9a47a4 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72050c6a2c1e45a88d0db4418995b455 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f59ad585f3614878bc4988bcbe95dbf0 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/194234e3a92449dd97e6e8105b02e9c2 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d03f27cf51a690ed7285a1c1147d7b2b27342002","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-10-28 02:15:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"873f2a9790a1e2a78fbf69fb5ba7da051e844668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-10-28 02:22:03.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3473ff13d0cd610074a22a97afa194a6fa570a77","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-10-28 02:23:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52d5256176758a42b2137450cc90d587c9d63aa6","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-10-28 02:25:25.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bfc3c68a1f214ae863876db77167df110a83db8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 02:31:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a4678c3990441a7bebb1f0313502a7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36642ae0890b4064b6be2c57ae4cead0 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d139777c41aa4f6789c63380161fcd60 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c2338c2472842108d6fb003cb482a2f : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08f63f1e635449b6bfafa3a88c25bfc2 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"87250b84c2ec06878383af4e5e8acd28363d66e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 02:32:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78f518cb7b81e56a2e2f7b1c5f9a19bf8ea47e71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 02:40:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd3d64f3c8c4d3bac1e3347febaec8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60de4f92f93f49aaa9941db2ba844418 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3c2d0bf5d84e5bbd99a9bd07b1c510 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17faba849f2843789f16caf3214664ab : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a106d9d63aac46d993bb60bc723cb045 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"70b32b419a53e08e6184141e393ba8cb932d6434","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 02:40:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4bac1b845564b75765e1fc5a36b0538a3b939b81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 02:42:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79eae1b50624573b2063c9fe312b3cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff57b7a3757748f19a1e4f78ad47888f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83b1fb7d092fc7fc6a5694332066292f4888a468","revisions":{"8cde9e6769a9932902fa5bb370f87c8f0027466a":{"kind":"REWORK","_number":1,"created":"2022-10-27 10:21:11.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/92/862792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/92/862792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/92/862792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/92/862792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/92/862792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/92/862792/1"}}},"commit":{"parents":[{"commit":"c193e7e12ae5f23400b0f8dec430802f3fb925c1","subject":"Add a use case for the test unit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/c193e7e12ae5f23400b0f8dec430802f3fb925c1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-10-27 10:20:34.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-10-27 10:20:45.000000000","tz":480},"subject":"Add default conf for es","message":"Add default conf for es\n\nChange-Id: I141e0b249af692a4c46a80577251a6a33cb54c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8cde9e6769a9932902fa5bb370f87c8f0027466a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8cde9e6769a9932902fa5bb370f87c8f0027466a"}]},"branch":"refs/heads/master"},"83b1fb7d092fc7fc6a5694332066292f4888a468":{"kind":"REWORK","_number":2,"created":"2022-10-28 02:22:03.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/92/862792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/92/862792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/92/862792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/92/862792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/92/862792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/92/862792/2"}}},"commit":{"parents":[{"commit":"c193e7e12ae5f23400b0f8dec430802f3fb925c1","subject":"Add a use case for the test unit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/c193e7e12ae5f23400b0f8dec430802f3fb925c1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-10-27 10:20:34.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2022-10-28 02:22:02.000000000","tz":0},"subject":"Add default conf for es","message":"Add default conf for es\n\nChange-Id: I141e0b249af692a4c46a80577251a6a33cb54c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/83b1fb7d092fc7fc6a5694332066292f4888a468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/83b1fb7d092fc7fc6a5694332066292f4888a468"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
