)]}'
{"id":"openstack%2Fvenus~862886","triplet_id":"openstack%2Fvenus~master~Iea328d91fa8296cb24c5bdcbd851e944dc5f9b38","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-11-14 02:42:53.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-11-14 01:17:30.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-11-14 02:09:47.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}},"hashtags":[],"change_id":"Iea328d91fa8296cb24c5bdcbd851e944dc5f9b38","subject":"Fix pbr version check","status":"MERGED","created":"2022-10-28 12:08:56.000000000","updated":"2022-11-14 02:44:24.000000000","submitted":"2022-11-14 02:42:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862886","meta_rev_id":"098bc326991bcdd97dac692ebbef399e939c1592","_number":862886,"virtual_id_number":862886,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 02:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":2,"date":"2022-11-14 01:17:30.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-14 02:09:47.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-14 02:09:47.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-28 12:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-10 13:40:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-11-14 01:17:30.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"109097313752d3c1758cfd576a3d731b73afb6a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-28 12:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11a173b6275edb952f9ccfc58d62c6f548da58c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 12:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313368a902ae45fb8d3a6369d54fd6b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171f146c25734becb5c8c653c9032f44 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f4ca169f094c37a13d31e1b6fab2bf : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d1d28a31d3247fcbba057dfce53f980 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d7ed452ca984667870e93f42ada770f : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f1ef5324043a96ec94b3a4435940456d30aea774","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-11-14 01:17:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"847dd38cb8eb8251f8fd977ba2c434a4fa17bd6e","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2022-11-14 02:09:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbb93ef8d7c554f76969e0e57420cdf3d8ead548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 02:11:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d696d3ae6da9ebba5f28e2e6b66bcbc05f75653e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 02:42:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a78e42e9ea49fc81e1dc87bb352502\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99550b253d234db9a35167128cdce549 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e03b88acbecb4368aa7d64cde084bff6 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6291f82ed294d0f81e604eab6678f70 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d79e33e39e54e8aa22b8f7030d5e95c : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4abe5c4af8d4d9f6340681928b4ed6a5979105b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 02:42:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"098bc326991bcdd97dac692ebbef399e939c1592","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 02:44:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d096dadb753e4ef3bd2503606fe146a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d03f34c3ed4f178107f6925d9af916 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"665230ea54697fced42dfaf0ecf1c8e07b2053d4","revisions":{"665230ea54697fced42dfaf0ecf1c8e07b2053d4":{"kind":"REWORK","_number":1,"created":"2022-10-28 12:08:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/86/862886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/86/862886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/86/862886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/86/862886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/86/862886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/86/862886/1"}}},"commit":{"parents":[{"commit":"83b1fb7d092fc7fc6a5694332066292f4888a468","subject":"Add default conf for es","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/83b1fb7d092fc7fc6a5694332066292f4888a468"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-28 11:53:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-10-28 12:08:54.000000000","tz":120},"subject":"Fix pbr version check","message":"Fix pbr version check\n\nCurrently pbr version check uses \u0027venus\u0027 as package name, while\npython package name is openstack-venus (as per setup.cfg).\n\nThis results in following when running venus-api or venus-manager:\n\nTraceback (most recent call last):\n  File \"/var/lib/kolla/venv/lib/python3.10/site-packages/pbr/version.py\", line 475, in _get_version_from_importlib_metadata\n    distribution \u003d importlib_metadata.distribution(self.package)\n  File \"/usr/lib/python3.10/importlib/metadata/__init__.py\", line 957, in distribution\n    return Distribution.from_name(distribution_name)\n  File \"/usr/lib/python3.10/importlib/metadata/__init__.py\", line 548, in from_name\n    raise PackageNotFoundError(name)\nimportlib.metadata.PackageNotFoundError: No package metadata was found for venus\n\nChange-Id: Iea328d91fa8296cb24c5bdcbd851e944dc5f9b38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/665230ea54697fced42dfaf0ecf1c8e07b2053d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/665230ea54697fced42dfaf0ecf1c8e07b2053d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[]}
