)]}'
{"id":"openstack%2Fvenus~863306","triplet_id":"openstack%2Fvenus~master~Icc9721e5c46afe7d98be0ec67cb446e97edeef8a","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-11-11 02:07:34.000000000","reason":"\u003cGERRIT_ACCOUNT_31412\u003e replied on the change","reason_account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-11-14 01:05:27.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-11-14 00:58:44.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"Icc9721e5c46afe7d98be0ec67cb446e97edeef8a","subject":"Add A unit test case to search_params.","status":"MERGED","created":"2022-11-10 06:25:07.000000000","updated":"2022-11-14 01:06:25.000000000","submitted":"2022-11-14 01:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"863306","meta_rev_id":"873d5176daa3239a948443be46c332acf44f458a","_number":863306,"virtual_id_number":863306,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 01:05:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-11-11 05:38:30.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-11-14 00:58:44.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-11-14 00:58:44.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-10 06:27:39.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-11-10 06:27:39.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2022-11-10 06:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85723e625b7168554d6ec65c2d0d7e90b16c71a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-10 06:25:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"697e5bef9d36f6420823345e7e970efe453cbb1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 06:29:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67eb57a127a94dc4989dfce08e1172cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb4ad7d1a6c4c48af16260c97debc34 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d582e4ff79d641129e62015f803dbf58 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d631bdb360774c4e8088abe52cc458fc : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2708edc894184509bbf395e859a8a9eb : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4cc4e82ffe60136307767f5be6cccf892437cf87","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-11 02:07:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70abca517d896068a183f7915a1c611b8296e27d","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-11 02:15:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0461be6222f15f31090db2e6db317efde01d9e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-11 02:22:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c8f77ff27ffa82f5d376d0bfce11695e7d31035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 02:30:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e400332d860540e1b2c75c7641a824b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084b39e73de04bcb837f540eb0cd1899 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d50f4941259c4f799e9f7e16cbc45ac6 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae7c10a4032e40bbbd88eb45751f62dc : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4a62c0c2e6b4975bf22f57928d690cc : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"bb5cc9381f1d72cba75cd1ea615c52b5b1b39561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-11 02:37:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"304e8ed81d3f8caf63dc1e1b5c9969e90f462ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 02:50:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e64d6b9583244ba28627a685a5391d9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e9e000902a433eac3421ae1dab5629 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a615e43b5b7245388679170cc4689329 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b95c6efdb943f28cdb9f43322618d4 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81066092296045fbb8331fe99a68f41a : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0821e568956ab4403f23ceacef95f3c1c0d576eb","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-11 05:38:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d5909aba886156b2425e774568ef262bfc5ffdac","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-11-14 00:58:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"05dae7570704e5490f0d1898b11e97354a8f7447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 00:59:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ba995928256363fda36eccb0afb7d61675b64fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:05:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/125a59c3d7984e94af111404da8fd1bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709778f0509f49a3b46866fd53ac9e14 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b00b4d185f34eea82617799d0056210 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b2593f1c228439ba39c686a41dd7bb0 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13fec60af1dd4294abc87aa1ffccab5e : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"c8616724a83df86e0fe7184019494f9202afb1eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:05:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"873d5176daa3239a948443be46c332acf44f458a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:06:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898a2e225e45478ba2a212e671ae31a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996c835a4d3d4b669d559a921716d13f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ddd8761f779659f789246cbaa5163796eee33d2f","revisions":{"52daf8660cce551989a989c56276db4f16e721cc":{"kind":"REWORK","_number":1,"created":"2022-11-10 06:25:07.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/06/863306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/06/863306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/06/863306/1"}}},"commit":{"parents":[{"commit":"16ac2c94d92871cf59d0800d4e14b4f65a33af32","subject":"Add unit test for controller api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/16ac2c94d92871cf59d0800d4e14b4f65a33af32"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-10 06:21:26.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-10 06:21:26.000000000","tz":480},"subject":"Add A unit test case to search_params.","message":"Add A unit test case to search_params.\n\nChange-Id: Icc9721e5c46afe7d98be0ec67cb446e97edeef8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/52daf8660cce551989a989c56276db4f16e721cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/52daf8660cce551989a989c56276db4f16e721cc"}]},"branch":"refs/heads/master"},"22518a2334eb25cab8397db4162d8a596479f388":{"kind":"REWORK","_number":2,"created":"2022-11-11 02:22:12.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/06/863306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/06/863306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/06/863306/2"}}},"commit":{"parents":[{"commit":"16ac2c94d92871cf59d0800d4e14b4f65a33af32","subject":"Add unit test for controller api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/16ac2c94d92871cf59d0800d4e14b4f65a33af32"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-10 06:21:26.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-11 02:21:30.000000000","tz":480},"subject":"Add A unit test case to search_params.","message":"Add A unit test case to search_params.\n\nChange-Id: Icc9721e5c46afe7d98be0ec67cb446e97edeef8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/22518a2334eb25cab8397db4162d8a596479f388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/22518a2334eb25cab8397db4162d8a596479f388"}]},"branch":"refs/heads/master"},"ddd8761f779659f789246cbaa5163796eee33d2f":{"kind":"REWORK","_number":3,"created":"2022-11-11 02:37:14.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/06/863306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/06/863306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/863306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/06/863306/3"}}},"commit":{"parents":[{"commit":"16ac2c94d92871cf59d0800d4e14b4f65a33af32","subject":"Add unit test for controller api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/16ac2c94d92871cf59d0800d4e14b4f65a33af32"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-10 06:21:26.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-11 02:36:47.000000000","tz":480},"subject":"Add A unit test case to search_params.","message":"Add A unit test case to search_params.\n\nChange-Id: Icc9721e5c46afe7d98be0ec67cb446e97edeef8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ddd8761f779659f789246cbaa5163796eee33d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ddd8761f779659f789246cbaa5163796eee33d2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[]}
