)]}'
{"id":"openstack%2Fvenus~863772","triplet_id":"openstack%2Fvenus~master~I5424c2ec3c3b1fe4bf1bbdb4f4f6dee6578de0a8","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-11-14 01:03:28.000000000","reason":"Change was submitted"},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-11-14 01:03:28.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-11-14 00:58:34.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2022-11-14 01:03:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5424c2ec3c3b1fe4bf1bbdb4f4f6dee6578de0a8","subject":"Add unit test for controller api","status":"MERGED","created":"2022-11-06 04:00:06.000000000","updated":"2022-11-14 01:04:26.000000000","submitted":"2022-11-14 01:03:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863772","meta_rev_id":"d0b41d70e24165009152b6fbd13756b79fc4186d","_number":863772,"virtual_id_number":863772,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 01:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-11-11 02:07:15.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-11-14 00:58:34.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-11-14 00:58:34.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-06 04:01:28.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"state":"REVIEWER"},{"updated":"2022-11-06 04:01:28.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-11-06 04:01:28.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2022-11-06 04:05:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4177a72f5fc6c31b0803c0f35d9cf23cc1e513c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-06 04:00:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c14ff9606b62cfbcf6ff4a047608fa2ce6569b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-06 04:05:45.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c30e5d8c42d42149061997681abbfa0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dded931eebc4321ab8448e15d0d62fb : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0779fe2da7f45b189055f2bb1973d85 : FAILURE in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/224482228b3e4241a5eb6ecf960f4f2c : FAILURE in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d05a9ef4c3114889a697bf34f6e16895 : FAILURE in 3m 44s\n\nWarning:\n  Comments left for invalid file venus/tests/unit/fakes.py","accounts_in_message":[],"_revision_number":1},{"id":"ae2101189cbe23323bc1155f843be012a7044369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-07 01:09:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c67d03422541e3ef07451455f4dcfe3fb77a5970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-07 01:14:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"47e8913eceee22e4b31b6b77ee9b2c9143de584c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 01:20:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/669693f2b6074bf1b5a6a3c89c0c41f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf2ff72b0db4560b78f856f77c19bd8 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c31224fe8e4bc884f1f7c93fa237d6 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1f2efacadcf4e53bd7eb108be0324db : FAILURE in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7eb762b65b84a68812af4c8f217c752 : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2d19d9ada265542f983437f4938f760b68232ba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-10 05:27:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1714392196ab4aa108fa80dbb8daf5a4b1c2b29b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 05:31:39.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1174393d7294ca1a7772c9bb49c86b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50481a4909904c32b456601a7c604529 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a32182619284a99a98114f1c58129d4 : FAILURE in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4ef1f3277e6413090f194a6659a54d6 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0158032aaf8401ea96c86a373e684bf : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":4},{"id":"2c5639f0ab4dd01a6f36d381833c26f8ef6ce845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-10 06:08:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"19f6c1dd4cc3afd5cef86176d5f039b4c257d0f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 06:13:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f410cca46fc429eb37c2cc6ff06fe1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c746b77d3a634278bcc0373a3ffa7016 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a4969b3cbc0471aa162fc604a68aa4f : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abeeed0c241e4c1d95af47e2d5bd14d7 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14cc4639e31b489da6ab71b86e2730d3 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"94c6c9f90c42778bbc55090f523b144cae0c7773","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-11 02:07:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ab9eb0a3599e7abe379ab1cf16828f8af9be9759","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-11-14 00:58:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b117287a800c0391d24e205804a535384af85584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 00:58:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9253dc98aa36c609462ac7dd3824e52ecec3c104","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:03:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffa4224381024e6ba44ce47eef74a4f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2684b0606a4f0d860209da9cac8d43 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb52afce97c428b992206a60b41f020 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e904ec49c10145c782a761bb5d2a1e7c : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b11daa410c84583b5ba66aea6dce40b : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":5},{"id":"a472e8e6cdc08f691e7491bb0fe5a1a986a3624b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:03:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d0b41d70e24165009152b6fbd13756b79fc4186d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:04:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d79114f7ffe54f7798b19510cf4e2977\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fa262b39df49a29087cdbceb2b2eeb : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"16ac2c94d92871cf59d0800d4e14b4f65a33af32","revisions":{"a8d1c84202ff257849a676790dc2c9b2708b9e96":{"kind":"REWORK","_number":1,"created":"2022-11-06 04:00:06.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/72/863772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/72/863772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/72/863772/1"}}},"commit":{"parents":[{"commit":"716f2fb4ca2782333f3b9181dfb4af0e2ca38227","subject":"Add fake request for test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/716f2fb4ca2782333f3b9181dfb4af0e2ca38227"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-11-06 03:59:53.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-11-06 03:59:53.000000000","tz":480},"subject":"Add unit test for controller api","message":"Add unit test for controller api\n\nChange-Id: I5424c2ec3c3b1fe4bf1bbdb4f4f6dee6578de0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a8d1c84202ff257849a676790dc2c9b2708b9e96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a8d1c84202ff257849a676790dc2c9b2708b9e96"}]},"branch":"refs/heads/master"},"385a21377e30da07599cde56242e4c379325da4c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-07 01:09:24.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/72/863772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/72/863772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/72/863772/2"}}},"commit":{"parents":[{"commit":"93184e1e25df17639bf657ad66ec797dabeb0e9e","subject":"Add fake request for test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/93184e1e25df17639bf657ad66ec797dabeb0e9e"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-11-06 03:59:53.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2022-11-07 01:09:24.000000000","tz":0},"subject":"Add unit test for controller api","message":"Add unit test for controller api\n\nChange-Id: I5424c2ec3c3b1fe4bf1bbdb4f4f6dee6578de0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/385a21377e30da07599cde56242e4c379325da4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/385a21377e30da07599cde56242e4c379325da4c"}]},"branch":"refs/heads/master","description":"Rebase"},"47f76584f10f97d267c1382809166dae9e4efbb8":{"kind":"REWORK","_number":3,"created":"2022-11-07 01:14:05.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/72/863772/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/72/863772/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/72/863772/3"}}},"commit":{"parents":[{"commit":"93184e1e25df17639bf657ad66ec797dabeb0e9e","subject":"Add fake request for test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/93184e1e25df17639bf657ad66ec797dabeb0e9e"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-11-06 03:59:53.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2022-11-07 01:14:05.000000000","tz":0},"subject":"Add unit test for controller api","message":"Add unit test for controller api\n\nChange-Id: I5424c2ec3c3b1fe4bf1bbdb4f4f6dee6578de0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/47f76584f10f97d267c1382809166dae9e4efbb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/47f76584f10f97d267c1382809166dae9e4efbb8"}]},"branch":"refs/heads/master"},"3473277120ab38ed51acba6ead59b25fc92cb902":{"kind":"REWORK","_number":4,"created":"2022-11-10 05:27:25.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/72/863772/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/72/863772/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/72/863772/4"}}},"commit":{"parents":[{"commit":"93184e1e25df17639bf657ad66ec797dabeb0e9e","subject":"Add fake request for test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/93184e1e25df17639bf657ad66ec797dabeb0e9e"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-11-06 03:59:53.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-10 05:26:48.000000000","tz":480},"subject":"Add unit test for controller api","message":"Add unit test for controller api\n\nChange-Id: I5424c2ec3c3b1fe4bf1bbdb4f4f6dee6578de0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3473277120ab38ed51acba6ead59b25fc92cb902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3473277120ab38ed51acba6ead59b25fc92cb902"}]},"branch":"refs/heads/master"},"16ac2c94d92871cf59d0800d4e14b4f65a33af32":{"kind":"REWORK","_number":5,"created":"2022-11-10 06:08:53.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/72/863772/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/72/863772/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/72/863772/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/72/863772/5"}}},"commit":{"parents":[{"commit":"93184e1e25df17639bf657ad66ec797dabeb0e9e","subject":"Add fake request for test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/93184e1e25df17639bf657ad66ec797dabeb0e9e"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2022-11-06 03:59:53.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-10 06:08:32.000000000","tz":480},"subject":"Add unit test for controller api","message":"Add unit test for controller api\n\nChange-Id: I5424c2ec3c3b1fe4bf1bbdb4f4f6dee6578de0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/16ac2c94d92871cf59d0800d4e14b4f65a33af32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/16ac2c94d92871cf59d0800d4e14b4f65a33af32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[]}
