)]}'
{"id":"openstack%2Fvenus~864277","triplet_id":"openstack%2Fvenus~master~I7475fdec93262e0b9783526a90e1c706cffc00d5","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-11-14 10:31:20.000000000","reason":"Change was submitted"},"34375":{"account":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"last_update":"2022-11-14 10:31:20.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-11-14 10:20:29.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}},"hashtags":[],"change_id":"I7475fdec93262e0b9783526a90e1c706cffc00d5","subject":"Add a unit test case \u0027invalid_index_type\u0027.","status":"MERGED","created":"2022-11-14 07:58:03.000000000","updated":"2022-11-14 10:32:58.000000000","submitted":"2022-11-14 10:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"864277","meta_rev_id":"563075465d272751a20d30df5697e8801dcde063","_number":864277,"virtual_id_number":864277,"owner":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-14 10:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-11-14 09:28:58.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-11-14 10:20:29.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-11-14 10:20:29.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-14 07:58:32.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-11-14 07:58:32.000000000","updated_by":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2022-11-14 08:02:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a18e35a1b6fc79459c10f2e7250daac5d3173211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-14 07:58:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c446c11e05561b71ea9441e30a74a50e86be42a0","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-14 08:00:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"739e171c4fa494ee3cc7387f0935651cda5b6d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 08:02:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d26cd3a3a024f0893acfd42277c3b47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6665aab04c343f5b86b81d76c193168 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d5bd7b2fed43e59ba520a64588fc82 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ade39e8b9c71424d970117df7d0479bb : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/805f03c67cec43a2ae906b188ecbf268 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7f679e9c62072236b6a7986bdf8bce63a2289186","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"date":"2022-11-14 08:10:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da442f6fe97354049b37b92d83c1c3f11c1c6022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 08:11:15.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/venus for 864276,4","accounts_in_message":[],"_revision_number":2},{"id":"9ddd96856046f599d5115e7abd7fe00751b889c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-14 08:14:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5017dc183123478195c85b01f44f20d9ca6c3528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 08:18:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55f67373d8ef473abbf5f40799378740\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6b4c66eb274f6987923b2035db5a9d : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9178a4422767420ab4d8896e9c6f14b7 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77fe3b289a7c4dc58bdda4186c05a0bf : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2619482e4fc94d1d8d967b7e9de4f3f0 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"df9553f3a76df7431c75a4e0b16cb8e9babbb384","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-11-14 09:28:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"21992ea712724bbbd48b11572ff7e7bc61c27b61","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-11-14 10:20:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f744468a8c55cc237288df24c127cace523c3ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 10:20:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fef3bdeb59366ae9003d519803dde3ff9476219c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 10:31:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7744e0f5656482bbc3ebf1afc50cd29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe51c51d1cd480883890ffb28121758 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b53e0c3ecc44f9f971e493e532bf703 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8899524937724281858b8cf5096ff834 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4dfa3f7f341c45999640253bf96906e1 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c6006aeeb0c86c1ac01fca38d46d677e3a2dfb5d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 10:31:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"563075465d272751a20d30df5697e8801dcde063","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 10:32:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/154cfd7ce2c7415e9bdd736db50ab326\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed77e01e25d4bb1886b72fe0c789e55 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ab0d345de37ee0749ffbfc0eb0287a45c83878df","revisions":{"a045a7967733a6d314b93ae5a9309091bb013856":{"kind":"REWORK","_number":1,"created":"2022-11-14 07:58:03.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/77/864277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/77/864277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/77/864277/1"}}},"commit":{"parents":[{"commit":"68b4aaf02f3290ce78ac540c9175ba99cda311b9","subject":"Supplement a unit test case \u0027valid_index_type\u0027.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/68b4aaf02f3290ce78ac540c9175ba99cda311b9"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-14 07:57:03.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-14 07:57:03.000000000","tz":480},"subject":"Add a unit test case \u0027invalid_index_type\u0027.","message":"Add a unit test case \u0027invalid_index_type\u0027.\n\nChange-Id: I7475fdec93262e0b9783526a90e1c706cffc00d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a045a7967733a6d314b93ae5a9309091bb013856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a045a7967733a6d314b93ae5a9309091bb013856"}]},"branch":"refs/heads/master"},"9e8de62f35eb26882f4c9098412a76bccccc03fd":{"kind":"REWORK","_number":2,"created":"2022-11-14 08:10:57.000000000","uploader":{"_account_id":34375,"name":"hejunli","email":"hejunli-s@inspur.com","username":"hejunli","status":"working"},"ref":"refs/changes/77/864277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/77/864277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/77/864277/2"}}},"commit":{"parents":[{"commit":"68b4aaf02f3290ce78ac540c9175ba99cda311b9","subject":"Supplement a unit test case \u0027valid_index_type\u0027.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/68b4aaf02f3290ce78ac540c9175ba99cda311b9"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-14 07:57:03.000000000","tz":480},"committer":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-14 08:10:35.000000000","tz":480},"subject":"Add a unit test case \u0027invalid_index_type\u0027.","message":"Add a unit test case \u0027invalid_index_type\u0027.\n\nChange-Id: I7475fdec93262e0b9783526a90e1c706cffc00d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/9e8de62f35eb26882f4c9098412a76bccccc03fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/9e8de62f35eb26882f4c9098412a76bccccc03fd"}]},"branch":"refs/heads/master"},"ab0d345de37ee0749ffbfc0eb0287a45c83878df":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-14 08:14:39.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/77/864277/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/77/864277/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/77/864277/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/77/864277/3"}}},"commit":{"parents":[{"commit":"9adc236bc1302596912426f3bf6af3f6521d509c","subject":"Supplement a unit test case \u0027valid_index_type\u0027.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/9adc236bc1302596912426f3bf6af3f6521d509c"}]}],"author":{"name":"hejunli","email":"hejunli-s@inspur.com","date":"2022-11-14 07:57:03.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2022-11-14 08:14:39.000000000","tz":0},"subject":"Add a unit test case \u0027invalid_index_type\u0027.","message":"Add a unit test case \u0027invalid_index_type\u0027.\n\nChange-Id: I7475fdec93262e0b9783526a90e1c706cffc00d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ab0d345de37ee0749ffbfc0eb0287a45c83878df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ab0d345de37ee0749ffbfc0eb0287a45c83878df"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[]}
