)]}'
{"id":"openstack%2Fvenus~872070","triplet_id":"openstack%2Fvenus~master~I892b7954bd220c862972d975cd57700654664610","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2023-01-29 03:10:50.000000000","reason":"\u003cGERRIT_ACCOUNT_30455\u003e replied on the change","reason_account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}},"hashtags":[],"change_id":"I892b7954bd220c862972d975cd57700654664610","subject":"set check delete es index time as 120s","status":"MERGED","created":"2023-01-29 02:10:57.000000000","updated":"2023-01-29 03:45:32.000000000","submitted":"2023-01-29 03:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872070","meta_rev_id":"829b8418a623863c8890e20215f549705f813dc0","_number":872070,"virtual_id_number":872070,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-29 03:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2023-01-29 02:15:37.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":2,"date":"2023-01-29 03:10:50.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2023-01-29 02:15:37.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":1,"date":"2023-01-29 03:10:50.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-29 02:15:37.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"},{"updated":"2023-01-29 02:43:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b2edfb2c22c77a414e28ef7d3e53c8de4974afe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-01-29 02:10:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f43ed65dadc104cff1ed268ee6a6aa163e3dd36a","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2023-01-29 02:15:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cba894cc1f913aac6774246e3d241f14ee02d0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 02:43:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fecdb20848f4b35a98bc857d4a6737c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d0ea1b867b40118bc7f862daa681b0 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2458595460804f92813a2fda9d58f196 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3eef800c83444fdbafa484289ed9ad8 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c520852a416340368dc61ddd956c38ca : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bad1580104f3b3f82539f6b9cd0ca6d9f44f623b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 02:43:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7954ca14e8e01cfac99070d9f1ffc4afc4c77ae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 02:49:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a085dfeb0652403a8fb0d4fc9e5236a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c97f142ca63461f938bb2d1d1cac13f : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a711ca90cea841f5914487df6371e3cf : POST_FAILURE in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91f6b8f0e8024e628a772597686385d1 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e98b4b317634ac181b52109ce35ac3d : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e41ab26a9a3d3d090b947a97bc993507193414c6","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-01-29 03:10:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7970d4858ceb443181f6e0e5d5afed1235cb8d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 03:20:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d17c6362d2c43de989b1d30c131bab3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b2e28e720bf46b486d2f2a0bccc2263 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2233f89e2e4710915d949aea8480ce : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e89d22f0b1540fd9ba6a06f0f1e992a : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fb88591d183431aaa0209d0adb50011 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"df57253dea34914ec4d8917df7efa9fd5fa46422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 03:21:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a223fc786b85bcec462900bb70529fe272920e81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 03:44:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2185fb29e2d4e4c9f4e1b95041aed71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4ee097f55c445aab017c8a4ef4e267 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae42c31b88e4435877bb36027471ca5 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05341d56b177433e90df0ac9d7b7d36c : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e3c81195e3b43728c78477134da0bb6 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8e4148fd30f07c7e278bc7c7be5051e43e27abeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 03:44:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"829b8418a623863c8890e20215f549705f813dc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 03:45:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ed0831ea304c2a9a547d3d1da57750\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390291330cd342b0ac5f43620c7447b7 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36b3e3469ca904a1ae230f4fe0c18d160151e5ee","revisions":{"36b3e3469ca904a1ae230f4fe0c18d160151e5ee":{"kind":"REWORK","_number":1,"created":"2023-01-29 02:10:57.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/70/872070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/70/872070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/70/872070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/70/872070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/70/872070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/70/872070/1"}}},"commit":{"parents":[{"commit":"7eda95ccd3c418f2957eba480d8b0b0794e9789e","subject":"Merge \"Fix tox4 error for venus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/7eda95ccd3c418f2957eba480d8b0b0794e9789e"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-01-29 02:10:43.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-01-29 02:10:43.000000000","tz":480},"subject":"set check delete es index time as 120s","message":"set check delete es index time as 120s\n\nChange-Id: I892b7954bd220c862972d975cd57700654664610\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/36b3e3469ca904a1ae230f4fe0c18d160151e5ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/36b3e3469ca904a1ae230f4fe0c18d160151e5ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
