)]}'
{"id":"openstack%2Fvenus~872071","triplet_id":"openstack%2Fvenus~master~I355b9b062f0424292e371817c82c02f19f3ebe5e","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2023-01-29 07:46:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I355b9b062f0424292e371817c82c02f19f3ebe5e","subject":"log output after judge it","status":"MERGED","created":"2023-01-29 07:12:32.000000000","updated":"2023-01-29 07:46:56.000000000","submitted":"2023-01-29 07:46:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872071","meta_rev_id":"105d9da8c3ae5fa031e267fd2d99ebad511fe20f","_number":872071,"virtual_id_number":872071,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-29 07:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2023-01-29 07:34:59.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2023-01-29 07:34:59.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-29 07:24:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-29 07:34:59.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"26b8e2b62fbf3ef44e9a8bac41ece089925378c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-01-29 07:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5438aa0bcd2b41924628ac9a178fa44217bb923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 07:24:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a72dfd01dcf48dfa358fdd1af39e0fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc6978c8c2e42c8adca7968235acfc8 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b8d183069441e286e87573c6adf03f : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e8ac49831f846f882241ef441a8c888 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a051f2d382f3438c8291eabbd590b6b9 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"62bf284ec3f39c72a618d7c8574d562e9bc2d54c","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2023-01-29 07:34:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e508c26b83379e574a584e96830fc0f8e85c651f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 07:35:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad380f76e97a5d05de6c09531134c01dcb64bb1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 07:46:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027a1e40db4e4fa59fb9121b36d9f417\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fe0a6b377d4c38b4157eab9b53530f : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308e14ff993d4b1ca54fe70733520995 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/361713e8402d40c085d4bac4e1899f08 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5963f53ab48f446c9af4123ee766739d : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"82fa7b195260ee28c86eba75652b73d12848d746","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 07:46:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"105d9da8c3ae5fa031e267fd2d99ebad511fe20f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 07:46:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/759225c84b8a45ab8421e54469a90a2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ee1047f65b480b8bf74d37ebfc86c3 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf49e2878a2645a5e7ccbfa8f5cc1dd7ed622bc3","revisions":{"bf49e2878a2645a5e7ccbfa8f5cc1dd7ed622bc3":{"kind":"REWORK","_number":1,"created":"2023-01-29 07:12:32.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/71/872071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/71/872071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/71/872071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/71/872071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/71/872071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/71/872071/1"}}},"commit":{"parents":[{"commit":"36b3e3469ca904a1ae230f4fe0c18d160151e5ee","subject":"set check delete es index time as 120s","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/36b3e3469ca904a1ae230f4fe0c18d160151e5ee"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-01-29 07:12:22.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-01-29 07:12:22.000000000","tz":480},"subject":"log output after judge it","message":"log output after judge it\n\nChange-Id: I355b9b062f0424292e371817c82c02f19f3ebe5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bf49e2878a2645a5e7ccbfa8f5cc1dd7ed622bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bf49e2878a2645a5e7ccbfa8f5cc1dd7ed622bc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}}]}],"submit_requirements":[]}
