)]}'
{"id":"openstack%2Fvenus~873053","triplet_id":"openstack%2Fvenus~master~Iefa91fe14b9cc2e2a96d31469e75fd316d9d1a22","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2023-02-08 07:38:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iefa91fe14b9cc2e2a96d31469e75fd316d9d1a22","subject":"change oslo context version","status":"ABANDONED","created":"2023-02-08 02:50:22.000000000","updated":"2023-02-08 07:38:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e28d742368e59e07f9e4baa231c31d778241591","_number":873053,"virtual_id_number":873053,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-08 05:58:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":-2,"date":"2023-02-08 07:38:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":-1,"date":"2023-02-08 07:38:35.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-08 03:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-08 07:38:35.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"7a98d2c89105c0ee43b74a83e61354e19066ded8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-08 02:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1058cb58701be8ff8776b37c213c575f4cbb9cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 03:10:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/710ccb7230914485bb152c828bccf1ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031b138942bf423b8ed6870858f83ec3 : FAILURE in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0f8879558f4cca8fd15e3034918925 : FAILURE in 2m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3af7a09caed143fd874207c835f31928 : FAILURE in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ee84c880b16435694ad610a816a9712 : FAILURE in 2m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7aa61c67d3cce317d718a35f50c56c58f8770d7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-08 05:44:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"efb1a97e8901b7e2600f9079f122742e5cf85235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 05:58:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f54a0634e638466dbc6057ee0eae57c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ace72f989974e6fb6d25b4ba54e186a : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d290014ad01f45a880b9ce56d2f3b065 : FAILURE in 2m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/252fe996a9ac4b5497aad81e3d3e7c48 : FAILURE in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c237c19707a49dab1cc68bba1f61f72 : FAILURE in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a22309632a71ebb70945954c83a3d2f617fcc4eb","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2023-02-08 07:38:35.000000000","message":"Patch Set 2: Code-Review-2 Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"6e28d742368e59e07f9e4baa231c31d778241591","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2023-02-08 07:38:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d8309d22319cc726cd234c1b0c3689fc8e8d051","revisions":{"3cb32396698f24717e7b1a0e145f8fcf40443f51":{"kind":"REWORK","_number":1,"created":"2023-02-08 02:50:22.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/53/873053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/53/873053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/53/873053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/53/873053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/53/873053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/53/873053/1"}}},"commit":{"parents":[{"commit":"78d18716c424f9fc2e64b7fd8c88075a276bd000","subject":"all index output to log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/78d18716c424f9fc2e64b7fd8c88075a276bd000"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-02-08 02:50:06.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-02-08 02:50:06.000000000","tz":480},"subject":"change oslo context version","message":"change oslo context version\n\nChange-Id: Iefa91fe14b9cc2e2a96d31469e75fd316d9d1a22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3cb32396698f24717e7b1a0e145f8fcf40443f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/3cb32396698f24717e7b1a0e145f8fcf40443f51"}]},"branch":"refs/heads/master"},"8d8309d22319cc726cd234c1b0c3689fc8e8d051":{"kind":"REWORK","_number":2,"created":"2023-02-08 05:44:59.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/53/873053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/53/873053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/53/873053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/53/873053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/53/873053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/53/873053/2"}}},"commit":{"parents":[{"commit":"78d18716c424f9fc2e64b7fd8c88075a276bd000","subject":"all index output to log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/78d18716c424f9fc2e64b7fd8c88075a276bd000"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-02-08 02:50:06.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-02-08 05:44:44.000000000","tz":480},"subject":"change oslo context version","message":"change oslo context version\n\nChange-Id: Iefa91fe14b9cc2e2a96d31469e75fd316d9d1a22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8d8309d22319cc726cd234c1b0c3689fc8e8d051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8d8309d22319cc726cd234c1b0c3689fc8e8d051"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
