)]}'
{"id":"openstack%2Fvenus~874636","triplet_id":"openstack%2Fvenus~master~Ie303ce31e4809a2816de135c69cc0d3e14d88432","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-02-22 08:33:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie303ce31e4809a2816de135c69cc0d3e14d88432","subject":"add api:set custom config","status":"MERGED","created":"2023-02-22 08:00:56.000000000","updated":"2023-02-22 08:34:56.000000000","submitted":"2023-02-22 08:33:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874636","meta_rev_id":"27b8f7042d7f93167ed71cb899d7e357b5de06b6","_number":874636,"virtual_id_number":874636,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 08:33:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-02-22 08:22:55.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-02-22 08:22:55.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 08:02:24.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2023-02-22 08:06:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0183f5c7ccfae18bfb49a211423d2b893dd1334","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-22 08:00:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c4a34df486800cc8c8c5adeb0b101fbe76dc5a4","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-22 08:02:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b169718418e6b9f0159dc934c300219be2e6885f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:06:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efa8dfb30f344f8da6664efdbe39e80c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6c4603180f43eb821a17e66fb33acf : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23d6fd0b1d644a69897a6d3b9bd7669f : FAILURE in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1abba8c62c2c480287f0c35db33836d3 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e314d4be6924d6b949acf6722fea7ee : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"43d1db84dd0759a3839a1dc8a7a0d45f4237acb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-22 08:11:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16f49b6db5a9614a9f65f8dcb8b4b81d8e244754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:21:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a106d00b629648789bf68f5ac7304d4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497b7c7317104ec48270b1e22eb4102c : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821e283cc4dc47288a6908500b85798c : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/075bb0f8aa114905924d1a9a2d649660 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fcd68bb18cd4f14909396cb490ff515 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"027f2c56cd33c80f6cb3507c767c298088ffd4a7","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-22 08:22:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17149a29315de25599ef75ea248a112c53e0ae09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:23:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87f0baa58e8c97d63b5406136eae59ad4e7e7700","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:33:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/181b6b8ca56346b6917676035945a7dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf5746de1ce4401a19fd639e3234910 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2297f693d347da8c2a9cda1cf5d5f0 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ab493b0af1449b3b76f2c75e128ae2c : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/145a02bd4b264c63a643cde83994b8ca : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"ec8446198bdc61b80f833813dd84f5c8499bc56a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:33:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"27b8f7042d7f93167ed71cb899d7e357b5de06b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:34:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c80f262f21a04b9e94ac8239c8911138\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ddb63fbee824b3cbb72a6f942b277ad : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d360c61ff2a271cbfcb261975c452eb63abfa937","revisions":{"636960fa3664170746a3e4c5c199ae5bd2b114a0":{"kind":"REWORK","_number":1,"created":"2023-02-22 08:00:56.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/36/874636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/36/874636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/36/874636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/36/874636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/36/874636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/36/874636/1"}}},"commit":{"parents":[{"commit":"8b6dc69e119c9717c230457e8716a4eafeeef081","subject":"rename custom config example file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8b6dc69e119c9717c230457e8716a4eafeeef081"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:00:46.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:00:46.000000000","tz":0},"subject":"add api:set custom config","message":"add api:set custom config\n\nChange-Id: Ie303ce31e4809a2816de135c69cc0d3e14d88432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/636960fa3664170746a3e4c5c199ae5bd2b114a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/636960fa3664170746a3e4c5c199ae5bd2b114a0"}]},"branch":"refs/heads/master"},"d360c61ff2a271cbfcb261975c452eb63abfa937":{"kind":"REWORK","_number":2,"created":"2023-02-22 08:11:48.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/36/874636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/36/874636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/36/874636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/36/874636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/36/874636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/36/874636/2"}}},"commit":{"parents":[{"commit":"8b6dc69e119c9717c230457e8716a4eafeeef081","subject":"rename custom config example file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8b6dc69e119c9717c230457e8716a4eafeeef081"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:00:46.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:09:47.000000000","tz":0},"subject":"add api:set custom config","message":"add api:set custom config\n\nChange-Id: Ie303ce31e4809a2816de135c69cc0d3e14d88432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d360c61ff2a271cbfcb261975c452eb63abfa937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d360c61ff2a271cbfcb261975c452eb63abfa937"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
