)]}'
{"id":"openstack%2Fvenus~874637","triplet_id":"openstack%2Fvenus~master~Ic6578595821ee9d30de72ccac6f003ce8a77b457","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-02-22 09:35:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6578595821ee9d30de72ccac6f003ce8a77b457","subject":"add set_costom_config api ref","status":"MERGED","created":"2023-02-22 08:44:47.000000000","updated":"2023-02-22 09:36:59.000000000","submitted":"2023-02-22 09:35:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874637","meta_rev_id":"1dbb6751e6dfe5e3b368a105abfa2a1b3ccde089","_number":874637,"virtual_id_number":874637,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 09:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-02-22 09:13:34.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-02-22 09:13:34.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 08:49:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 09:13:34.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"07c776928d865b031b3226f31a15334fc93896fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-22 08:44:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca2fcb4fa6865fb69a8158e15595b58d78e59535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:49:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c73e97ca6394136871f4e8b1efeea65\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44a90331744a496f9cf7250b02087654 : FAILURE in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d87d83c1aa4e0c889a617beafb2edc : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e9067a2b2b4bf2930a3693c35e4b17 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a9ee820a73a11c6138394607e100ec513c742245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-22 08:54:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f61ea23d185e1db6cf816f9164e3a4db2e51911b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:03:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5135f84c94fc412f9e99c7cc02aaecf3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dc2faffa2a4d4d638ea24997aea518e9 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536a8bb22bf143639b2c3bc1a34171ed : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40314d46244e435d918967fb78e6166a : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"defb5ab697587278856a1029ccb8bc0b0155a73b","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-22 09:13:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be577c4314cf12bb37f2f0eab881fbb2067b86cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:13:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ae19b866fbfb87e1986dbd45aea1092db2300973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:35:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da4620da03b4f26825414401317c7e1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e80324293e2948838efc4cf803bc8581 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a33e8c04784ae3a8041c006a5ea1c3 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b8100981334354a469755fea195d0a : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ee2d0360569e9b7dae7d976b0eb6249a5a03a758","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:35:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1dbb6751e6dfe5e3b368a105abfa2a1b3ccde089","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:36:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b28124b550ec428e8253a30136cde13a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a439687a973144c98057674776b165c3 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb641c9868148afb1a5365b6957adc1 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8e8fad8c67c72a8d6d608116bc7a49e73842991","revisions":{"b24359b3f0b329e6921a2fafc02dbadcb73fa3ad":{"kind":"REWORK","_number":1,"created":"2023-02-22 08:44:47.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/37/874637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/37/874637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/874637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/874637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/874637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/37/874637/1"}}},"commit":{"parents":[{"commit":"d360c61ff2a271cbfcb261975c452eb63abfa937","subject":"add api:set custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d360c61ff2a271cbfcb261975c452eb63abfa937"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:44:36.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:44:36.000000000","tz":0},"subject":"add set_costom_config api ref","message":"add set_costom_config api ref\n\nChange-Id: Ic6578595821ee9d30de72ccac6f003ce8a77b457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b24359b3f0b329e6921a2fafc02dbadcb73fa3ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b24359b3f0b329e6921a2fafc02dbadcb73fa3ad"}]},"branch":"refs/heads/master"},"a8e8fad8c67c72a8d6d608116bc7a49e73842991":{"kind":"REWORK","_number":2,"created":"2023-02-22 08:54:04.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/37/874637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/37/874637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/874637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/874637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/37/874637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/37/874637/2"}}},"commit":{"parents":[{"commit":"d360c61ff2a271cbfcb261975c452eb63abfa937","subject":"add api:set custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d360c61ff2a271cbfcb261975c452eb63abfa937"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:44:36.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-22 08:53:46.000000000","tz":0},"subject":"add set_costom_config api ref","message":"add set_costom_config api ref\n\nChange-Id: Ic6578595821ee9d30de72ccac6f003ce8a77b457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a8e8fad8c67c72a8d6d608116bc7a49e73842991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a8e8fad8c67c72a8d6d608116bc7a49e73842991"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
