)]}'
{"id":"openstack%2Fvenus~874659","triplet_id":"openstack%2Fvenus~master~Id49aeaeebcc4a480ca085a59ac5f4f35ed31e64a","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2023-06-08 08:02:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id49aeaeebcc4a480ca085a59ac5f4f35ed31e64a","subject":"add missing dependence of oslo_reports","status":"ABANDONED","created":"2023-02-21 16:25:43.000000000","updated":"2023-06-08 08:02:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af3560047c58f0f209e6ceab42c46563e0fbe8d8","_number":874659,"virtual_id_number":874659,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-21 16:33:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 16:33:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38558e7ec2b86fca942e868c5ea21152254afcdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-21 16:25:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eeaf9950cfe1c6b3a48e35d654825161d33de079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:33:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcbe8b4233eb40acbc64f66ddef2f6b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0dfb6fc0c3742a68e1bc63fefc6abb8 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/696effe07a34445a900ee42c058d851b : FAILURE in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76115914325546cb88d19db83a0c5d44 : FAILURE in 2m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/650db76db8124518b459a9a132023226 : FAILURE in 2m 10s","accounts_in_message":[],"_revision_number":1},{"id":"af3560047c58f0f209e6ceab42c46563e0fbe8d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-06-08 08:02:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5035b262f346af82e6192486f26da023ed44bad4","revisions":{"5035b262f346af82e6192486f26da023ed44bad4":{"kind":"REWORK","_number":1,"created":"2023-02-21 16:25:43.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/59/874659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/59/874659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/59/874659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/59/874659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/59/874659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/59/874659/1"}}},"commit":{"parents":[{"commit":"78d18716c424f9fc2e64b7fd8c88075a276bd000","subject":"all index output to log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/78d18716c424f9fc2e64b7fd8c88075a276bd000"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-02-21 16:22:59.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-02-21 16:22:59.000000000","tz":480},"subject":"add missing dependence of oslo_reports","message":"add missing dependence of oslo_reports\n\nChange-Id: Id49aeaeebcc4a480ca085a59ac5f4f35ed31e64a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5035b262f346af82e6192486f26da023ed44bad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5035b262f346af82e6192486f26da023ed44bad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
