)]}'
{"id":"openstack%2Fvenus~875630","triplet_id":"openstack%2Fvenus~master~I4fe02164980bb4a03dee023ee262bdf9c94e211d","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-02-28 08:10:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4fe02164980bb4a03dee023ee262bdf9c94e211d","subject":"Unified configuration name","status":"MERGED","created":"2023-02-28 07:37:37.000000000","updated":"2023-02-28 08:10:59.000000000","submitted":"2023-02-28 08:10:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875630","meta_rev_id":"ded9437833ac5fb7b79e2a408db47a0088d2acea","_number":875630,"virtual_id_number":875630,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 08:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-02-28 07:58:31.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-02-28 07:58:31.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 07:46:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 07:58:31.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"ea66f66345ea7e7d8a10915aca7eb0446c16f801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-28 07:37:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e0b90642bf54558ef90c0b8cb79efbafca51be6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 07:46:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de0adedc09fb4591af17ee769eabaa36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/503d83af69614793a0aada714435b3be : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b58e1b1e6ac54f20aaaa084b2bd8af0b : FAILURE in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb22b3303bdd44f39adf68f4f3696faf : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1037ddb3bd0f4778836a794a74eecd67 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5d8112f1eaeae41dff363d8c8af30aacfece4bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-28 07:49:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc47a4616608de7ecd73b9d5939a4a005022f20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 07:57:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b45867336bb4a099970f132a835825e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0331b30439254cef9291b744f9f08aea : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ff73c388c743839de954fa987f5f20 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f773fcf1100a4a71986081c251399a5d : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afc5d2abf3db47e1af62ff10cb51fc2f : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"55fa91d451232348d084129973fb88939e6c1256","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-28 07:58:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbe273a53a35b79a94e904e7cb7478ae025a30a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 07:58:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4dd9331905788bc622c0effef52299ed748824f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 08:10:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a2f65d7f244201b9646280e4516fff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09fc4d8136d64023a438248eeea4b574 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63af6ef97134abf9a862cf9da7f96f9 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7ab15dbcc7a4ee2bc96d34208532471 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c7b5e1ed2b14a0caa0ac468c8910212 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b3d3b333af1430539f9b960dd7d1c3579c3a2e6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 08:10:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ded9437833ac5fb7b79e2a408db47a0088d2acea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 08:10:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a6783f19bd4c9ab3ead59f8c071d6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9288067fc0bf46f481caebc7fc72ffa4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b3a21c330885081eed312d360771a6bce5d404c","revisions":{"98ba88134606e2fe89adbc3d2c7c87d77d380b27":{"kind":"REWORK","_number":1,"created":"2023-02-28 07:37:37.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/30/875630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/30/875630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/875630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/875630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/875630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/30/875630/1"}}},"commit":{"parents":[{"commit":"71a1ace424dee456d1a64e45a4337f4178de8a80","subject":"detail fixed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/71a1ace424dee456d1a64e45a4337f4178de8a80"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 07:37:21.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 07:37:21.000000000","tz":0},"subject":"Unified configuration name","message":"Unified configuration name\n\nChange-Id: I4fe02164980bb4a03dee023ee262bdf9c94e211d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/98ba88134606e2fe89adbc3d2c7c87d77d380b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/98ba88134606e2fe89adbc3d2c7c87d77d380b27"}]},"branch":"refs/heads/master"},"6b3a21c330885081eed312d360771a6bce5d404c":{"kind":"REWORK","_number":2,"created":"2023-02-28 07:49:20.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/30/875630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/30/875630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/875630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/875630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/30/875630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/30/875630/2"}}},"commit":{"parents":[{"commit":"71a1ace424dee456d1a64e45a4337f4178de8a80","subject":"detail fixed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/71a1ace424dee456d1a64e45a4337f4178de8a80"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 07:37:21.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 07:49:06.000000000","tz":0},"subject":"Unified configuration name","message":"Unified configuration name\n\nChange-Id: I4fe02164980bb4a03dee023ee262bdf9c94e211d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6b3a21c330885081eed312d360771a6bce5d404c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6b3a21c330885081eed312d360771a6bce5d404c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
