)]}'
{"id":"openstack%2Fvenus~875631","triplet_id":"openstack%2Fvenus~master~I9836ff8ebd1522601c57dd6e52c58a63914fa382","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-02-28 09:33:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9836ff8ebd1522601c57dd6e52c58a63914fa382","subject":"Unified log format","status":"MERGED","created":"2023-02-28 08:31:23.000000000","updated":"2023-02-28 09:34:30.000000000","submitted":"2023-02-28 09:33:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875631","meta_rev_id":"2a7c771a164ea4e76ff10b97bf765f51614b7982","_number":875631,"virtual_id_number":875631,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 09:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-02-28 09:19:08.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-02-28 09:19:08.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 08:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 09:19:08.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"111c65d16dbb55b13cc5f6cca7f91a08ebab7864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-28 08:31:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b42563ceaba45dd7bf19dceb0411bfa2c6e0e3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 08:51:35.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e58174a97c04cb2a58b9355fbae4bb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566dd6a96fcc4d2e95487fa15dada8b4 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7664d06c0c840d3ba8513413aa7f8f1 : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69d25f8e6bb49bf83356c9b8cee2121 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd9f2e3f52fa4b1cbaeabcfbf27b3e01 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f6a928cdf1ef3baf315dd5eda7e7b7bf15cdbfcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-28 08:54:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e49bbc8bb7a6e01abe514a75e7435e44d76f494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:16:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ed4daa936b42ac87353d449722a815\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f8dddbb4f943ab92a6076bd9239b70 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccaae9c22ac84e169e56c3a905d95b52 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a99f8587a8c44daafe41eb2ddf096d6 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b6b0462a1194f5b9873dcd6592449b1 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7b85e6d7233a84f415dc4ec0852c35c132a34da5","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-28 09:19:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f256387329283dcad47bf0a18c6871ce33e7813d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:20:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30840cc6c4bac260d889b15369cce3bda663bfdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:33:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b4d0f8b821342d692c606268798dde8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22bbac231ed84bf0b998f41ac3ea9780 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c8d47229df645bb9f11b7e7620cb0cb : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412899129dd7450abbd65db9c08842f2 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d241a6c8cda4f3882d8a462ce747130 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8b254996fb5f1b1df23af5d9172a3ea8a5998332","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:33:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2a7c771a164ea4e76ff10b97bf765f51614b7982","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:34:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88a3c90830fc4d9c81551172774d9b06\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215acb896e95403dae98856c098952bf : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8fe3950e0efe670d6ef550fee5eae323c47e09af","revisions":{"791fdfcb019a4e324fcc76ee396bfd679b296936":{"kind":"REWORK","_number":1,"created":"2023-02-28 08:31:23.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/31/875631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/31/875631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/31/875631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/31/875631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/31/875631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/31/875631/1"}}},"commit":{"parents":[{"commit":"6b3a21c330885081eed312d360771a6bce5d404c","subject":"Unified configuration name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6b3a21c330885081eed312d360771a6bce5d404c"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 08:30:37.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 08:30:37.000000000","tz":0},"subject":"Unified log format","message":"Unified log format\n\nChange-Id: I9836ff8ebd1522601c57dd6e52c58a63914fa382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/791fdfcb019a4e324fcc76ee396bfd679b296936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/791fdfcb019a4e324fcc76ee396bfd679b296936"}]},"branch":"refs/heads/master"},"8fe3950e0efe670d6ef550fee5eae323c47e09af":{"kind":"REWORK","_number":2,"created":"2023-02-28 08:54:56.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/31/875631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/31/875631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/31/875631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/31/875631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/31/875631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/31/875631/2"}}},"commit":{"parents":[{"commit":"6b3a21c330885081eed312d360771a6bce5d404c","subject":"Unified configuration name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6b3a21c330885081eed312d360771a6bce5d404c"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 08:30:37.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 08:54:39.000000000","tz":0},"subject":"Unified log format","message":"Unified log format\n\nChange-Id: I9836ff8ebd1522601c57dd6e52c58a63914fa382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8fe3950e0efe670d6ef550fee5eae323c47e09af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8fe3950e0efe670d6ef550fee5eae323c47e09af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
