)]}'
{"id":"openstack%2Fvenus~875635","triplet_id":"openstack%2Fvenus~master~I069a6c13cc4010ea24fbddcbe33fb38400cb8a67","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-02-28 12:02:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I069a6c13cc4010ea24fbddcbe33fb38400cb8a67","subject":"adjust the log level to avoid printing a large number of logs","status":"MERGED","created":"2023-02-28 11:20:02.000000000","updated":"2023-02-28 12:03:05.000000000","submitted":"2023-02-28 12:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875635","meta_rev_id":"3b69882d74c8d4bf27ff83b217c5bd0f202747d3","_number":875635,"virtual_id_number":875635,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 12:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-02-28 11:45:26.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-02-28 11:45:26.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 11:32:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 11:45:26.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"02802dc66e6733cc18ee3a30839cf36296cda969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-28 11:20:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d15d1726e778f05a3f03c0e8b23ddfb24c97f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:32:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a36c2a55cf03493bac711bfdc22e2bc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bfafa37bf8848528ca112bc8e590ed4 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a66e7a937354b82bdacf8df9be8320b : FAILURE in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1affed1c22fd4491a9f29983741a32fe : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c682ec5b5fc42bd97ae15990e65070a : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7695403f1a2447a1f8adfd459e6d0467528474b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-02-28 11:34:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5fa493e4f009960780338124420969f7b7e82112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:44:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1828e33bed74283ad99936139b0ccb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f0cab8d9ae4762b52a209d653c5a78 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fe4f62886c4facb8be278b49e539a3 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0dd8ba05b934e8287aacecb07d506c3 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f033599a57ab4a009bb4b8162d25f3b6 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b663a3dc724c99e66ee7f37b2506b5bb5a754d49","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-02-28 11:45:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dfea9c5755c7999adcaece7f4ae0ea05c5cded59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:45:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ffa892c4d373d63dded2734ca2155c23813fa4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:02:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95b4ff4e546e422480965c55b468f5d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d422181fbc84527a59e6275065942cf : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfbdfb6124b04f45bf2b3a01cda4698b : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caa4120c74d445629fc46eae8fce7903 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d5aa85cd06c41129f80268e0dbda201 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"3d5ea9884f4bc20f6477456fa979ed69cf3dbeaf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:02:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3b69882d74c8d4bf27ff83b217c5bd0f202747d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 12:03:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075a00d4cd2f4a31bbcd1c291c7f45c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19139bf82fd24e939333544da0772f93 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69414d4f72941cc60ada422488464e19d6eb64ee","revisions":{"477b15fb6dbeb691105c38b27f15f7ad0b930400":{"kind":"REWORK","_number":1,"created":"2023-02-28 11:20:02.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/35/875635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/35/875635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/875635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/875635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/875635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/35/875635/1"}}},"commit":{"parents":[{"commit":"41bb0f4e0314c85ad618fcee665f6f622f378961","subject":"fix details \u0026 optimize delete es index code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/41bb0f4e0314c85ad618fcee665f6f622f378961"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 11:19:44.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 11:19:44.000000000","tz":0},"subject":"adjust the log level to avoid printing a large number of logs","message":"adjust the log level to avoid printing a large number of logs\n\nChange-Id: I069a6c13cc4010ea24fbddcbe33fb38400cb8a67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/477b15fb6dbeb691105c38b27f15f7ad0b930400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/477b15fb6dbeb691105c38b27f15f7ad0b930400"}]},"branch":"refs/heads/master"},"69414d4f72941cc60ada422488464e19d6eb64ee":{"kind":"REWORK","_number":2,"created":"2023-02-28 11:34:31.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/35/875635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/35/875635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/875635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/875635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/35/875635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/35/875635/2"}}},"commit":{"parents":[{"commit":"41bb0f4e0314c85ad618fcee665f6f622f378961","subject":"fix details \u0026 optimize delete es index code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/41bb0f4e0314c85ad618fcee665f6f622f378961"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 11:19:44.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-02-28 11:34:20.000000000","tz":0},"subject":"adjust the log level to avoid printing a large number of logs","message":"adjust the log level to avoid printing a large number of logs\n\nChange-Id: I069a6c13cc4010ea24fbddcbe33fb38400cb8a67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/69414d4f72941cc60ada422488464e19d6eb64ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/69414d4f72941cc60ada422488464e19d6eb64ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[]}
