)]}'
{"id":"openstack%2Fvenus~877368","triplet_id":"openstack%2Fvenus~master~I30e56238a5504dc7643df518b7aecd4cbcd37e72","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2023-03-15 02:36:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I30e56238a5504dc7643df518b7aecd4cbcd37e72","subject":"fix post error","status":"MERGED","created":"2023-03-14 12:25:05.000000000","updated":"2023-03-15 02:38:04.000000000","submitted":"2023-03-15 02:36:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877368","meta_rev_id":"43f33577630caf4264ec9d1af985c276da95c336","_number":877368,"virtual_id_number":877368,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-15 02:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":2,"date":"2023-03-15 02:28:05.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"all":[{"value":1,"date":"2023-03-15 02:28:05.000000000","_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-14 12:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-15 02:28:05.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"}],"messages":[{"id":"bf8af10d1ceb886334532c60f7fbe619a22218d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-03-14 12:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"155f1c4697b891da44afa73b741a6f21572f1bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 12:33:46.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/838d6e5979b74df3b3b53fb147cfecfb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15a2ea2b7844dccaff636137aa68530 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6715d8f8e375407180af8489e7345cad : FAILURE in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca236623a4d34ddaab2554005699c223 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/082c7a2cdbe949a19bda1e687c8b1181 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fc3e94af654628726272c036ca96463670d36c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-03-15 01:26:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bd58434ca4f317bba98b974b1ab8e129c4a5423e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 01:36:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5077dd0ee9f44a0b801ae6851f39ed77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38da15d9f9954c2b8d760822d6b8370c : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498131d2ec6e407f807e30600fba9215 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9191a47933a949efbe1389c9d2201af8 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b636f89a2a047a59b147e986560468e : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"381f739179488d60b9d0cba5a35aa7aa965e5b11","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2023-03-15 02:28:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81ce7b97444e5aa025e14e3a1fce911426c4e017","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 02:28:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19786366908c3cf111912575ad41db616695ffd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 02:36:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15181863047445a49c8074ab457e30df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984414ebf4794fc0ad12f9131ecd580b : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c01e7ebdeefe462d8687dd6914e3c535 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3edacaca0b9a4a408f00159e7b00dbf7 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7056a00ff94b4ea988652e83d0ac4130 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c4c9f49c65b295635eabe2b19281e631186521e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 02:36:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"43f33577630caf4264ec9d1af985c276da95c336","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 02:38:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a0fa07895d49699ebed118012c163b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aaaf7382b934370b9d8346a2589688f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a29ab7ddbd1f023340ff527ea2f50dfabbf34a9","revisions":{"ccb124134afd2b2aed2f646fbd2b4b6eea36f405":{"kind":"REWORK","_number":1,"created":"2023-03-14 12:25:05.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/68/877368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/68/877368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/68/877368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/68/877368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/68/877368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/68/877368/1"}}},"commit":{"parents":[{"commit":"e83cbb7ccf97bdba7bbae58e29e941ca55f26293","subject":"add interface log storage days","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e83cbb7ccf97bdba7bbae58e29e941ca55f26293"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-03-14 12:24:51.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-03-14 12:24:51.000000000","tz":480},"subject":"fix post error","message":"fix post error\n\nChange-Id: I30e56238a5504dc7643df518b7aecd4cbcd37e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ccb124134afd2b2aed2f646fbd2b4b6eea36f405"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ccb124134afd2b2aed2f646fbd2b4b6eea36f405"}]},"branch":"refs/heads/master"},"8a29ab7ddbd1f023340ff527ea2f50dfabbf34a9":{"kind":"REWORK","_number":2,"created":"2023-03-15 01:26:16.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/68/877368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/68/877368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/68/877368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/68/877368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/68/877368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/68/877368/2"}}},"commit":{"parents":[{"commit":"e83cbb7ccf97bdba7bbae58e29e941ca55f26293","subject":"add interface log storage days","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e83cbb7ccf97bdba7bbae58e29e941ca55f26293"}]}],"author":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-03-14 12:24:51.000000000","tz":480},"committer":{"name":"jiasirui","email":"jiasirui@inspur.com","date":"2023-03-15 01:25:59.000000000","tz":480},"subject":"fix post error","message":"fix post error\n\nChange-Id: I30e56238a5504dc7643df518b7aecd4cbcd37e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8a29ab7ddbd1f023340ff527ea2f50dfabbf34a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8a29ab7ddbd1f023340ff527ea2f50dfabbf34a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
