)]}'
{"id":"openstack%2Fvenus~877912","triplet_id":"openstack%2Fvenus~master~I49531fb4206092c79f3cf7e2ba05dba14e4a82e1","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34934":{"account":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"last_update":"2023-03-20 09:31:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I49531fb4206092c79f3cf7e2ba05dba14e4a82e1","subject":"Remove extra placeholders","status":"MERGED","created":"2023-03-20 08:56:29.000000000","updated":"2023-03-20 09:32:34.000000000","submitted":"2023-03-20 09:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877912","meta_rev_id":"1c13dc34ef55fe8441195b44b012aebd88d62695","_number":877912,"virtual_id_number":877912,"owner":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 09:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-03-20 09:19:43.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-03-20 09:19:43.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 09:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 09:19:43.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"1244c99041e2e8b1563858fa169df2da3f6fc8a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2023-03-20 08:56:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0faf34d2da26c04b0a12904fa7b20545d265113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:07:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa545fd301a041368d20383f74e59f17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0174ac7443514293ba352a6d7e0acd4a : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c1a2d501084b999fb7ae3409e025d8 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc00b88d1aa743648ccb6c2fb0da9e36 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ab58c9e54b342368c07d62cc9d5e14e : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3e37e35f96de44313a58d2ff60db488ea1c0c08a","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-03-20 09:19:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adb92b5f427dd85032e290882c1be042806cbcd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:20:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cef647bc5b4bbbd33f0cec7f53b2ae5da94c6a1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:31:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb89cadbe1844fd78bf7e7264f53f234\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a448a11416f48c28c9fab8a4f440ab2 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1879aec21ac45ffb6a74c1d62523eab : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bbac39c64be4db1b13b0eb3e414c071 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5f048342cf24f15a2439bb723209757 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b30f7ad135bf177b8bbbe7e33795d20e39829c4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:31:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c13dc34ef55fe8441195b44b012aebd88d62695","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e414cd3a6aec449eb8caa82b884e29bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67faa91e3c824c399f80f0accb6db3b2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d63d4c50cdd44405a7f7b558baaed3f19531bd93","revisions":{"d63d4c50cdd44405a7f7b558baaed3f19531bd93":{"kind":"REWORK","_number":1,"created":"2023-03-20 08:56:29.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/12/877912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/12/877912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/12/877912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/12/877912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/12/877912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/12/877912/1"}}},"commit":{"parents":[{"commit":"4ec937bf354fa8ace05b0a37a51d503d2e8c4d0e","subject":"update version api sample","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4ec937bf354fa8ace05b0a37a51d503d2e8c4d0e"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-03-20 08:53:53.000000000","tz":480},"committer":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-03-20 08:53:53.000000000","tz":480},"subject":"Remove extra placeholders","message":"Remove extra placeholders\n\nChange-Id: I49531fb4206092c79f3cf7e2ba05dba14e4a82e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d63d4c50cdd44405a7f7b558baaed3f19531bd93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d63d4c50cdd44405a7f7b558baaed3f19531bd93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
