)]}'
{"id":"openstack%2Fvenus~877914","triplet_id":"openstack%2Fvenus~master~I904d72620886061a5ae52efe778360fb027bfeee","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34934":{"account":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"last_update":"2023-03-20 09:29:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I904d72620886061a5ae52efe778360fb027bfeee","subject":"The difference between python3 and python2 dependent packages","status":"ABANDONED","created":"2023-03-20 09:07:47.000000000","updated":"2023-03-20 09:29:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"79751bea25b34fe67c33ddd0cb9aaa6587e2975c","_number":877914,"virtual_id_number":877914,"owner":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-20 09:19:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 09:19:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fa6bd3d5c2b822cea559064e21bf4cc3b74dd6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2023-03-20 09:07:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96b98c35e6aa2b008ce594d45dede1188240edc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:19:53.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95d5266d962a4ba0880cb34ceca25657\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9726e100e74c948d8b6af6f7bccee1 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c120a2099b440298402dff22f271df0 : FAILURE in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a81fd05852ce465d8b0a79be49ab76ad : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9e6dbbe294a4fe58aef59963beca160 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"79751bea25b34fe67c33ddd0cb9aaa6587e2975c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2023-03-20 09:29:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7184e5950057d7e5e868a399a24e43774761dde","revisions":{"a7184e5950057d7e5e868a399a24e43774761dde":{"kind":"REWORK","_number":1,"created":"2023-03-20 09:07:47.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/14/877914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/14/877914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/877914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/877914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/14/877914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/14/877914/1"}}},"commit":{"parents":[{"commit":"d63d4c50cdd44405a7f7b558baaed3f19531bd93","subject":"Remove extra placeholders","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d63d4c50cdd44405a7f7b558baaed3f19531bd93"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-03-20 09:07:35.000000000","tz":480},"committer":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-03-20 09:07:35.000000000","tz":480},"subject":"The difference between python3 and python2 dependent packages","message":"The difference between python3 and python2 dependent packages\n\nChange-Id: I904d72620886061a5ae52efe778360fb027bfeee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a7184e5950057d7e5e868a399a24e43774761dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a7184e5950057d7e5e868a399a24e43774761dde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
