)]}'
{"id":"openstack%2Fvenus~877925","triplet_id":"openstack%2Fvenus~master~I24fdadd53da60109eadbc585aff880c6914e58c1","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34934":{"account":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"last_update":"2023-05-17 02:33:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24fdadd53da60109eadbc585aff880c6914e58c1","subject":"The variable is not declared in advance","status":"MERGED","created":"2023-03-21 03:31:06.000000000","updated":"2023-05-17 02:34:28.000000000","submitted":"2023-05-17 02:33:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877925","meta_rev_id":"6215691b4f191ad7af4a4b3f834084782ba0c886","_number":877925,"virtual_id_number":877925,"owner":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 02:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 02:10:45.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-17 02:10:45.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 03:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 02:10:45.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"e65ab2d1fed46bfd74279fad884dd797b5b4e63d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2023-03-21 03:31:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2935d212bcaf329506e5885102b8679425cea037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 03:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8451dc2f6b4439db1b2b9966c0c2dcf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab09a9ff2ce14833995ed834aac5d537 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18566478d1584a4093111ec1a27cc4c7 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6a404ccbf2f4c1982f47b6215f70486 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba953c0b1b414b82a85a5f792b4c03e3 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"705d34ad1a6428b7c00c5ce76a3ed90e969efa3a","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-05-17 02:10:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9668bb80748ab22fcdd68594d8f50841732cb079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 02:11:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ab19b45ba4cee86eaa61ec881830bd10f9d83fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 02:33:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/016d69b9ec444d6097c9ccd3ada8839a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bcfc49e90e4e7f8b0a0f03da374f74 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1875e700c3c443158e77bb96b1c1d87a : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a67d080a400466d9697e55e5e83047f : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f84be029e21440ca9f2ab2f56c0cc84f : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7a22bceaf424f23b54b9ace4519eccc : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"af0bd2a0d1b89a56b8d54d94d8538328f856acb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 02:33:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6215691b4f191ad7af4a4b3f834084782ba0c886","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 02:34:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c531c8d8e0f4d90a4d6244c5d322666\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff131cb5c5641a7b8019d5a662591e7 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f0fd87da6432e6fc6af60a213d309d3714d48f6","revisions":{"6f0fd87da6432e6fc6af60a213d309d3714d48f6":{"kind":"REWORK","_number":1,"created":"2023-03-21 03:31:06.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/25/877925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/25/877925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/25/877925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/25/877925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/25/877925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/25/877925/1"}}},"commit":{"parents":[{"commit":"d63d4c50cdd44405a7f7b558baaed3f19531bd93","subject":"Remove extra placeholders","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d63d4c50cdd44405a7f7b558baaed3f19531bd93"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-03-21 03:30:29.000000000","tz":480},"committer":{"name":"yangzhe","email":"17839227138@163.com","date":"2023-03-21 03:30:29.000000000","tz":480},"subject":"The variable is not declared in advance","message":"The variable is not declared in advance\n\nChange-Id: I24fdadd53da60109eadbc585aff880c6914e58c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6f0fd87da6432e6fc6af60a213d309d3714d48f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6f0fd87da6432e6fc6af60a213d309d3714d48f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
