)]}'
{"id":"openstack%2Fvenus~878688","triplet_id":"openstack%2Fvenus~master~If48734fc0223c5390baf63c7d1171a3d70be1d71","project":"openstack/venus","branch":"master","topic":"master1","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-05-17 02:33:19.000000000","reason":"Change was submitted"},"30409":{"account":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"last_update":"2023-05-17 02:33:19.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2023-05-17 02:33:19.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-05-17 02:11:33.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}},"hashtags":[],"change_id":"If48734fc0223c5390baf63c7d1171a3d70be1d71","subject":"remove not user variables","status":"MERGED","created":"2023-04-03 08:14:24.000000000","updated":"2023-05-17 02:33:19.000000000","submitted":"2023-05-17 02:33:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878688-master1","meta_rev_id":"bf440040672a261c673eb7e362fda5e5e8a35ddf","_number":878688,"virtual_id_number":878688,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 02:33:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 02:11:33.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-17 02:11:33.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 08:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-04 02:02:10.000000000","updated_by":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2023-05-04 02:02:10.000000000","updated_by":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2023-05-04 02:04:39.000000000","updated_by":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"196d20f547c5694795bd8663b56c5d422a873ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2023-04-03 08:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98d8a394a2a6fc6f84a4087d0cb2dc3e9d60de2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 08:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/367364243af040379fc6f13e76373f59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f695c1e6df4012bf61bd7aa4a19800 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614585ba3b2f43f8b2abe44629ba6743 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eacc17f8f6184aa5b4700ba2d25635af : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1c8a9ebc28945caa2117555192264e2 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b0daee0b49366a4dc44cf36fcbed6bf41ec305bb","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-05-17 02:11:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c6f1ed5a4a56547cfc5c6007fbfd434ef9af6bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 02:12:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6b920574c5dd6438cd3d06ea121720d83b9c304","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 02:33:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd96fc19e1f4df0b0943a8f491946f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd592abb704f42f1ab673d952581f1e1 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c04f670f4d749a5ace1103afb6ecbaf : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d16b0e7d745a499683115629200a7777 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0b3df74a8ed43f6a3c0df6f481c9318 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b57b978b8f38402483dd432cc0caa031 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"bf440040672a261c673eb7e362fda5e5e8a35ddf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 02:33:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"157fefdcabbfc126a47ed06d32e490b3850ed673","revisions":{"157fefdcabbfc126a47ed06d32e490b3850ed673":{"kind":"REWORK","_number":1,"created":"2023-04-03 08:14:24.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/88/878688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/88/878688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/88/878688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/88/878688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/88/878688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/88/878688/1"}}},"commit":{"parents":[{"commit":"d63d4c50cdd44405a7f7b558baaed3f19531bd93","subject":"Remove extra placeholders","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d63d4c50cdd44405a7f7b558baaed3f19531bd93"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2023-04-03 08:12:20.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2023-04-03 08:12:27.000000000","tz":480},"subject":"remove not user variables","message":"remove not user variables\n\nremove not use variables field for venus/mudules/search/action.py file\nChange-Id: I49a1c9a1b3d800ca13b73fa8894bf86115831f01\n\nChange-Id: If48734fc0223c5390baf63c7d1171a3d70be1d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/157fefdcabbfc126a47ed06d32e490b3850ed673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/157fefdcabbfc126a47ed06d32e490b3850ed673"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
