)]}'
{"id":"openstack%2Fvenus~885940","triplet_id":"openstack%2Fvenus~master~Iba4b68fcc8f997e0c02cf10882bbb10b32f96474","project":"openstack/venus","branch":"master","hashtags":[],"change_id":"Iba4b68fcc8f997e0c02cf10882bbb10b32f96474","subject":"fix  log store days as str","status":"MERGED","created":"2023-06-13 07:19:50.000000000","updated":"2023-06-13 08:12:27.000000000","submitted":"2023-06-13 08:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885940","meta_rev_id":"87fd2613fbdc050c08eca1ccdd0a4c5b8f9722d2","_number":885940,"virtual_id_number":885940,"owner":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-13 08:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-06-13 08:03:08.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-06-13 08:03:08.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 07:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd9de90514b7b1e1cba453c0b1fc6184cd1950d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-06-13 07:19:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"281771789e16aadf7f311cfac9df703f2507c000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 07:31:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb07d6238ece4ef1ad57f91e26d253fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ba431991634ddcbeb5b908bed7fd81 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11b6f2f80a93464b9d16e020dc2c9e71 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb4206d718994ff8b23271673e3adb3d : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/777c3700f69f455d94523936f707581e : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6c92eb303c14252b50970377c42f1af : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"36399c818ed65cd2a8cb03461ea4d5eadd5e5991","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-06-13 08:03:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f0f3a4dc081c9842a1091a29894a152fee8de7cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 08:03:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a78b2d351d7633aaeb0d43ab54d77e8e51db008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 08:11:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebf2309c3c84ade9b34b983514557e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53780c53eb1e4f7daa51118c12c9a298 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264de096c7eb4a91955c750e1a1b81d0 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c23918ee02714e05b627d073cccded25 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8c31984cac64d21b42aa002994b6ec8 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04d0241873a3400c90d3a3897a82f0b6 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0550b16bed43daac9b53ede0bea19d7930d39b02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 08:11:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87fd2613fbdc050c08eca1ccdd0a4c5b8f9722d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 08:12:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/545c91714c9a41e1b6fbf567b2176440\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26e750605b9476eb10758f6826c9fca : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"650ebbb2692e6771ef0e47220e3ffcc001961f1a","revisions":{"650ebbb2692e6771ef0e47220e3ffcc001961f1a":{"kind":"REWORK","_number":1,"created":"2023-06-13 07:19:50.000000000","uploader":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"ref":"refs/changes/40/885940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/40/885940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/885940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/885940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/40/885940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/40/885940/1"}}},"commit":{"parents":[{"commit":"aa9759b056026789db5b8d6b2d2331331637a840","subject":"fix operate error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/aa9759b056026789db5b8d6b2d2331331637a840"}]}],"author":{"name":"Frank Jia (贾思瑞)","email":"jiasirui@inspur.com","date":"2023-06-13 07:19:27.000000000","tz":480},"committer":{"name":"Frank Jia (贾思瑞)","email":"jiasirui@inspur.com","date":"2023-06-13 07:19:27.000000000","tz":480},"subject":"fix  log store days as str","message":"fix  log store days as str\n\nChange-Id: Iba4b68fcc8f997e0c02cf10882bbb10b32f96474\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/650ebbb2692e6771ef0e47220e3ffcc001961f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/650ebbb2692e6771ef0e47220e3ffcc001961f1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
