)]}'
{"id":"openstack%2Fvenus~889264","triplet_id":"openstack%2Fvenus~master~Ia8de8742de6f18394250ce4fe51da16d2048519b","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-07-24 09:55:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8de8742de6f18394250ce4fe51da16d2048519b","subject":"anomaly detect: add api (add rule)","status":"MERGED","created":"2023-07-24 09:01:03.000000000","updated":"2023-07-24 09:56:50.000000000","submitted":"2023-07-24 09:55:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889264","meta_rev_id":"dbb54239e63a411671e858812218ce213bd4b7b0","_number":889264,"virtual_id_number":889264,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 09:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-07-24 09:46:53.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-07-24 09:46:53.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-24 09:16:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-24 09:46:53.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"2ff202e4c62c1a602e66ccd732cb8cb2eadacded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 09:01:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a7c0dbe05751ff73c6198b3b125b9f2aee416e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 09:05:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f63e6106e1aa623e7015550f83dea196793e3a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:16:13.000000000","message":"Patch Set 2: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64cec396284a4eda8207299be22eb7ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfd01dc20664e2690cdf5a9daf1370d : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205d369e3af34ba18165712c289253e9 : FAILURE in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19b895dcc70e4d358d9dd92cc13df90c : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/366a54483533487091ca21b023bfd991 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e28812a99396417caf1767015fea997e : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"139f0999ffa43a59115a440a771edcc18011a7a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 09:22:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"881ee294eae6e5f2e29ea45e8afe96f426075203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:34:28.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68663af26d44427f863c305b095d47bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a1260f10b442808db98fb20835c2df : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed58095fca847ffa32d7784ee0c366d : FAILURE in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/016872967f38411eb9706c59e6e3eb6b : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33148de12f134531b559b151c39d7cee : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/346e4ab929a44b5a9b479497c2cd924a : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2b349756be253206f6cba54df67f5e3afc3a56af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 09:36:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c22b0f85cf2a93851892852783c695adefecb41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:45:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c6b1a827f844a8a399c02044c15ac5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6f0b591b924046bc364874e0cc1f44 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a80e8c4360a40a99cfd258eebc66dd3 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e4ea928dae43518bbf30fb7a786055 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f75da21da7d844ab8b23782cdace2eba : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95f64c3958ee4925af2f9b3582ef26ab : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"4d03b3c7086f1c5ded019bb424b89bd7e795b32e","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-07-24 09:46:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f1d78964ff4ef5823599690e788955401446080b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:47:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"43f44bdd84e1a27762f7574ea1299be3f7d7d60c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:55:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/870c8a1533d64c1f8738557cdd1a3b2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f33362be6f45e59f38f78af58066a7 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15287d3d531548c4ae85075acbf61920 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/840a0bf3fab74bd3846e9050d750ab12 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/403d0740661e4d958d09d727f723c001 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc300a5bd9864e5590dfc32bf414357d : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":4},{"id":"0ff21957a79711c7b5ad37db7051d0df87a726a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:55:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dbb54239e63a411671e858812218ce213bd4b7b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 09:56:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c597d504d144c728f9897a696b071d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71343a152f204bb2b232a7d12b9cdeb7 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8a79f8ee4068100b2527b0e74b3ba60778a2c234","revisions":{"4f4afee50ecab77cb8888ff6c3f0fc45c6a88657":{"kind":"REWORK","_number":1,"created":"2023-07-24 09:01:03.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/64/889264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/64/889264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/64/889264/1"}}},"commit":{"parents":[{"commit":"6c9e74770f7446ba903bc33a40cbeae5ce4b66b3","subject":"add anomaly detect db models","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6c9e74770f7446ba903bc33a40cbeae5ce4b66b3"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 08:59:38.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 08:59:38.000000000","tz":0},"subject":"anomaly detect: add api (add rule)","message":"anomaly detect: add api (add rule)\n\nChange-Id: Ia8de8742de6f18394250ce4fe51da16d2048519b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4f4afee50ecab77cb8888ff6c3f0fc45c6a88657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4f4afee50ecab77cb8888ff6c3f0fc45c6a88657"}]},"branch":"refs/heads/master"},"8120edb7619bdbe430581ce6b4c30723be3b07c1":{"kind":"REWORK","_number":2,"created":"2023-07-24 09:05:30.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/64/889264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/64/889264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/64/889264/2"}}},"commit":{"parents":[{"commit":"6c9e74770f7446ba903bc33a40cbeae5ce4b66b3","subject":"add anomaly detect db models","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6c9e74770f7446ba903bc33a40cbeae5ce4b66b3"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 08:59:38.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 09:05:10.000000000","tz":0},"subject":"anomaly detect: add api (add rule)","message":"anomaly detect: add api (add rule)\n\nChange-Id: Ia8de8742de6f18394250ce4fe51da16d2048519b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8120edb7619bdbe430581ce6b4c30723be3b07c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8120edb7619bdbe430581ce6b4c30723be3b07c1"}]},"branch":"refs/heads/master"},"52856d56cfa3392d293b0ba15cf08c84732eb658":{"kind":"REWORK","_number":3,"created":"2023-07-24 09:22:37.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/64/889264/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/64/889264/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/64/889264/3"}}},"commit":{"parents":[{"commit":"6c9e74770f7446ba903bc33a40cbeae5ce4b66b3","subject":"add anomaly detect db models","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6c9e74770f7446ba903bc33a40cbeae5ce4b66b3"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 08:59:38.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 09:22:22.000000000","tz":0},"subject":"anomaly detect: add api (add rule)","message":"anomaly detect: add api (add rule)\n\nChange-Id: Ia8de8742de6f18394250ce4fe51da16d2048519b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/52856d56cfa3392d293b0ba15cf08c84732eb658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/52856d56cfa3392d293b0ba15cf08c84732eb658"}]},"branch":"refs/heads/master"},"8a79f8ee4068100b2527b0e74b3ba60778a2c234":{"kind":"REWORK","_number":4,"created":"2023-07-24 09:36:30.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/64/889264/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/64/889264/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/64/889264/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/64/889264/4"}}},"commit":{"parents":[{"commit":"6c9e74770f7446ba903bc33a40cbeae5ce4b66b3","subject":"add anomaly detect db models","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/6c9e74770f7446ba903bc33a40cbeae5ce4b66b3"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 08:59:38.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 09:36:18.000000000","tz":0},"subject":"anomaly detect: add api (add rule)","message":"anomaly detect: add api (add rule)\n\nChange-Id: Ia8de8742de6f18394250ce4fe51da16d2048519b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8a79f8ee4068100b2527b0e74b3ba60778a2c234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8a79f8ee4068100b2527b0e74b3ba60778a2c234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
