)]}'
{"id":"openstack%2Fvenus~889265","triplet_id":"openstack%2Fvenus~master~I0e19db135e6fc602bd13e0db7528a8bcf2051520","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2023-07-24 10:30:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30455\u003e replied on the change","reason_account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-07-24 10:41:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e19db135e6fc602bd13e0db7528a8bcf2051520","subject":"anomaly detect: add api (delete rule)","status":"MERGED","created":"2023-07-24 10:13:15.000000000","updated":"2023-07-24 10:42:48.000000000","submitted":"2023-07-24 10:41:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889265","meta_rev_id":"3274026a382727739b0d55a906ef6de070cf4374","_number":889265,"virtual_id_number":889265,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 10:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-07-24 10:30:24.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-07-24 10:30:24.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-24 10:14:59.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"},{"updated":"2023-07-24 10:23:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"851547be88439438834bc1635f741223a8e7624b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 10:13:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"209426cf36d18e593c6189ca9d2eb5f47aa2f326","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-07-24 10:14:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"224aa84d25752a2e0da4e177c0a3da5185f42c3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 10:16:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3b0d3837ed1aac6b2b5c8f525c367e0863d28409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 10:23:20.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f55dbf6ccf784631bdfc06d6431d7517\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90ff85be1f643ec9497f75ac9b89f5f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa19d1d8a3ea4d01ae29603025601a0d : FAILURE in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b982b1550a0249bdb96792d2e75610cc : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f8b8d49352846bba12c2f5d2498d42a : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90f65fe357274e45ab04395e78887cf4 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ba141078e681feef75c46305c29e43b6443dfbb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 10:25:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b6081c05699f40fb9949e3acfd6a6d3beed1fc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 10:30:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/608e86935308401eb9c919ff6ed0a5be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee520bdde0846608c4095dd7bae7fc5 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b18f35834245f489214418a3ae3870 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56d5d256f6f14d7bb0c315392ea03c4a : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eeb933fac614965b894a949b24c9457 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba8cb6d188a14e9a861bab99fda86d29 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"ece2a8fcd3bebc182a9437ebe84dffe4c46dfc7f","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-07-24 10:30:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c26d4f33ebc965622d9214275a260971b1708aa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 10:30:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"64acb3fe003b64479c7c5547a6f783156ff42d9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 10:41:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71db0b8415474c5989a88996bf78a7d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd084f8a19424b73bf1451f4f47d9bae : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23ef781057b425fa6323d21c08400bb : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9f2ce9cf194bb1a55b11b501956199 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c615ed6cbcf4d34be9d1a7a9b635c8a : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f4549be8b6a4b748c41b49dbf081570 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":3},{"id":"534c0d245e47c753896f0f214ef439f63d18b88b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 10:41:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3274026a382727739b0d55a906ef6de070cf4374","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 10:42:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/449fc6f6e2554c6f812f702359ee32ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2edf842540b401ab7b82dd0fd6b5d56 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba0a023d9fb8c398703b1d0c047e51007bceefeb","revisions":{"45fb1f47ce985f7b3de164aa24f22b24164e485b":{"kind":"REWORK","_number":1,"created":"2023-07-24 10:13:15.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/65/889265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/65/889265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/65/889265/1"}}},"commit":{"parents":[{"commit":"8a79f8ee4068100b2527b0e74b3ba60778a2c234","subject":"anomaly detect: add api (add rule)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8a79f8ee4068100b2527b0e74b3ba60778a2c234"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:12:34.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:12:34.000000000","tz":0},"subject":"anomaly detect: add api (delete rule)","message":"anomaly detect: add api (delete rule)\n\nChange-Id: I0e19db135e6fc602bd13e0db7528a8bcf2051520\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/45fb1f47ce985f7b3de164aa24f22b24164e485b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/45fb1f47ce985f7b3de164aa24f22b24164e485b"}]},"branch":"refs/heads/master"},"dc89e4c2b6a628ea02858c63843d33b73d1ad3db":{"kind":"REWORK","_number":2,"created":"2023-07-24 10:16:05.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/65/889265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/65/889265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/65/889265/2"}}},"commit":{"parents":[{"commit":"8a79f8ee4068100b2527b0e74b3ba60778a2c234","subject":"anomaly detect: add api (add rule)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8a79f8ee4068100b2527b0e74b3ba60778a2c234"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:12:34.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:15:52.000000000","tz":0},"subject":"anomaly detect: add api (delete rule)","message":"anomaly detect: add api (delete rule)\n\nChange-Id: I0e19db135e6fc602bd13e0db7528a8bcf2051520\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dc89e4c2b6a628ea02858c63843d33b73d1ad3db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dc89e4c2b6a628ea02858c63843d33b73d1ad3db"}]},"branch":"refs/heads/master"},"ba0a023d9fb8c398703b1d0c047e51007bceefeb":{"kind":"REWORK","_number":3,"created":"2023-07-24 10:25:15.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/65/889265/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/65/889265/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/65/889265/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/65/889265/3"}}},"commit":{"parents":[{"commit":"8a79f8ee4068100b2527b0e74b3ba60778a2c234","subject":"anomaly detect: add api (add rule)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8a79f8ee4068100b2527b0e74b3ba60778a2c234"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:12:34.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:25:03.000000000","tz":0},"subject":"anomaly detect: add api (delete rule)","message":"anomaly detect: add api (delete rule)\n\nChange-Id: I0e19db135e6fc602bd13e0db7528a8bcf2051520\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ba0a023d9fb8c398703b1d0c047e51007bceefeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ba0a023d9fb8c398703b1d0c047e51007bceefeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
